)]}'
{"id":"openstack%2Fpuppet-neutron~957636","triplet_id":"openstack%2Fpuppet-neutron~master~Ibf260264d83f285008155223ada30423dc26254e","project":"openstack/puppet-neutron","branch":"master","topic":"remove-q-svc","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-19 10:15:35.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ibf260264d83f285008155223ada30423dc26254e","subject":"Remove support for standalone neutron-server","status":"MERGED","created":"2025-08-18 01:11:52.000000000","updated":"2025-08-19 11:38:33.000000000","submitted":"2025-08-19 11:37:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957636-remove-q-svc","meta_rev_id":"26eaf55c32b4114dde5dd76d742fa6e4d68919d9","_number":957636,"virtual_id_number":957636,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 11:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-08-18 14:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-08-19 07:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-08-19 10:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 04:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-18 14:40:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-08-19 07:03:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e40929af0824ff9b416aa54f63e905694e992b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 01:11:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b137c7722887d8d1fd3b49c0246958d93e0e0c16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 01:44:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f172bf6bb5842a18374b415376565cfc63594db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 01:47:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fcf8b6f7f133a033b5f32e97ed8f3640006d470a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 02:34:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0fc64924cfc15c80674dfc57b45e88b0b94909c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 04:57:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/471a89c1529d450384176ad3052dad01\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9bc78c68f0242a6901d1c175188bcfe : SUCCESS in 3m 40s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8696fa93f704c83bfb45b5b14d98781 : SUCCESS in 2m 48s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb9eb393332847f08b82980da22e6974 : FAILURE in 11m 08s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9c70123be8e4f479bfd57c186e86ede : FAILURE in 4m 54s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4113926b4a2d4ceca41a38af828aba8e : FAILURE in 8m 38s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c29e54372ca43cfb4d7f3f6f4589fd5 : SUCCESS in 23m 50s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92d096175f1b4491876c7af6448a5a45 : SUCCESS in 30m 49s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdfc86160fef4c44834f26b58a5a8f5f : FAILURE in 31m 40s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63abd158dc2e4c8d85f90923006264ed : SUCCESS in 44m 47s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2734a38e33cf42c8a02f8e25e22c5f79 : FAILURE in 16m 00s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/467bf71ce5cf475e995f52eed6ba668e : SUCCESS in 50m 11s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed5603f9a4d34f449bc2e1820dc6d426 : SUCCESS in 1h 06m 21s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1547b04014d94dc7b75087c2898759f8 : FAILURE in 31m 51s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c72212c7adb4104899a529badaa6eb8 : SUCCESS in 52m 37s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0e643a98edf4cefa791fd05ad8d48c0 : SUCCESS in 50m 12s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/86729a90759d4ee789818b3d3546a591 : SUCCESS in 11m 39s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bd6763dd8904208b7756bdfac477f9d : SUCCESS in 14m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10e9708daf6f44e0bffcc5e8cdefbca0 : SUCCESS in 13m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c46c1d1b08c4977bd52bb459f40e85d : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":4},{"id":"ea190ea25083855e846b7e7646142cad968a6cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 08:53:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fb3b0a9b52bead6a61588ae4fdccf4983be85ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 10:13:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fd55b319bec4477af7d2a0a492b2382\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aacc86763c694cd7aebe22a55d1d7674 : SUCCESS in 1m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7010639e225448bf814ecccdd47a8304 : SUCCESS in 2m 32s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/187254477e1048ff9ed7651f8892bc31 : FAILURE in 6m 09s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/596e6dd81232452ba27ef620c0f1592c : FAILURE in 4m 13s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceaab29a48e6439a9c62c4da3699b2a7 : FAILURE in 4m 42s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcbb71fd6d8c4fcf90a2e654aa811c88 : SUCCESS in 28m 31s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fca85b5fff3407ab165cca8d6d3be2f : SUCCESS in 38m 46s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65b6445cb41947ff8f4bb83a4334efa6 : FAILURE in 16m 09s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335dce4e2b4647dcaac4df807c92c988 : SUCCESS in 26m 28s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba4f146ae01a452b8d00f789da3b2d02 : FAILURE in 14m 43s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e9042de609b4ea5b05f34b96821eef6 : SUCCESS in 33m 56s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5555aeea0b54d34a7d04bcf7596113e : SUCCESS in 42m 42s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3bf3bb7bb5840c3850cb77056b35980 : FAILURE in 39m 51s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3654e7b5d89a4d31934dcac023ee4479 : SUCCESS in 47m 12s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3e3aaab1b5242d78af2667386f39e8a : SUCCESS in 56m 14s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/443cd0591d874c67a029ff6d2717efe4 : SUCCESS in 11m 38s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3220bceb4034140ba4409e94d607c72 : SUCCESS in 6m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c87846b99ae4e82bba5796fd10d7d79 : SUCCESS in 9m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a23a18411334e7a98b34e752e62e529 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":5},{"id":"5a59ad0b6ef3669b74086e0fd7420eb73f8bc930","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 11:19:02.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0218344a2030a97b3b92bfd8b697160be0c75f3e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 11:21:12.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"3576c912e2c3821de99ad4e739db23a26a51eeb2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 11:23:41.000000000","message":"Topic wip removed","accounts_in_message":[],"_revision_number":6},{"id":"da5a1ed1544a2930c395005f76a8682529701f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 11:23:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5b0c9f502978c50a17c08feb4a6628bf9369f03f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 11:23:55.000000000","message":"Topic wip removed","accounts_in_message":[],"_revision_number":7},{"id":"201cc23324d6e9612492957737e36b7ba7f70457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:38:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce499825ccc4464f89f4a880fc2e9138\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5517020b6f6c4326a886a47e7f67b326 : SUCCESS in 2m 09s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60c88befed134888b5ef0e8025a102ca : SUCCESS in 2m 18s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/deb34f66db784020b02800cf992d3a2f : SUCCESS in 12m 59s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4575a3e04fe476391ca9e9242b069dc : SUCCESS in 4m 30s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c425be7b3204ab0a229db81202fded3 : SUCCESS in 9m 16s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1566538c2d7745fdbaf50227ec5824d7 : SUCCESS in 1h 01m 57s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90e34754416645019987642b2f5f6a13 : SUCCESS in 56m 00s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5498a9b51d8942d6befd6a15e82e9e74 : FAILURE in 16m 01s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/077b0e7c7f464047b7110ce027291bc0 : SUCCESS in 30m 44s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa1fdf9b68ff4b7c90014647234b9bfd : FAILURE in 25m 35s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ed1b2a5c0b5422f9d8e68db8ca3ebce : SUCCESS in 31m 25s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f88cfddff9442cc92014131bda85ec1 : SUCCESS in 1h 07m 54s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46000b46accc48e3b5d46c04aea047bb : FAILURE in 17m 18s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c678daafc078424fbaf1f885d8d33399 : SUCCESS in 50m 02s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/876dd59d25ce4a078d89c09044e1f4d8 : SUCCESS in 56m 51s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a8ae1b043bb4d30aa65e50b394278a4 : SUCCESS in 9m 04s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f44c509e6c194539a40f8dfe25b2adf8 : SUCCESS in 11m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6e015bd3c2f4b62a71e65e5226cfdc7 : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9590e15beb4b4dcc823d1e656d863602 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":7},{"id":"719f4e8ed840fc462d47cb8025448bfbe84f6be3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 12:50:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bbf959b29713a37cf933b6bd654f12c2d6583ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 14:20:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"eeb0c59f5c5ed24acf91cf422b5cf12bbfb3b60a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-18 14:40:43.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3b7c9ca06866197c6e776d06e59fd415cd771a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:09:18.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957585 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"875a4cd6c92b6f9bd63a1ddf4c7c1116148eb33e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:09:28.000000000","message":"Topic remove-server removed","accounts_in_message":[],"_revision_number":8},{"id":"95994271272b86e1bb0be3a0039f5ef5311c8daf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:09:32.000000000","message":"Topic set to remove-q-svc","accounts_in_message":[],"_revision_number":8},{"id":"607de6351f137948668c5ade958d0fd14774d7ac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:09:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"23322a241e135f4dcc21e5054ae50432585854e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 17:07:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f1968d6f4fa49be9cfc81f5c3187484\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a385897aaedd47858636c43be1e5719a : SUCCESS in 3m 22s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64c9cdf83bde464fb11c19d266c5bb46 : SUCCESS in 1m 56s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f73dec5934045e19566befa70deeaa7 : SUCCESS in 12m 28s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d69ab39ec554cfdb6c990694ebdb42e : SUCCESS in 5m 15s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba6fd6986d634a5cb917b8e2d5ef7127 : SUCCESS in 7m 47s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a48af1bc5374c718528b9d0fce9d6eb : SUCCESS in 34m 10s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db1be46dd4e94a15b95dae57450fb291 : SUCCESS in 38m 49s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de9c9c571ff445fa971b5e245ca8cdce : SUCCESS in 27m 38s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8c34f522f2f4e518eb0a638f0f70396 : FAILURE in 21m 57s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26c100d1521347f782c0bf8159767e7f : FAILURE in 35m 39s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43ca570ced1f4d149859a94fd868c387 : SUCCESS in 58m 59s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/533fd2a217f346babfd003f125f320d1 : SUCCESS in 1h 12m 41s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/13abcdce12e74b96bef053553efed9a1 : SUCCESS in 54m 22s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/064e6c8e6a164cec8d13f16ca8208612 : SUCCESS in 49m 13s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6eaaf5f92f2d4a979350b5c9f43bc49a : SUCCESS in 32m 31s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/555876806d7c4b48bc4b79204eab0f33 : SUCCESS in 18m 43s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acd04573cffd47c192ef08880849d3bb : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35849f7cb74b4360b10d11d2abb7b9ef : SUCCESS in 10m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/091f939fa83b40bd926f7b6d4dd25054 : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":8},{"id":"f179bf7cfc1b2f42ecd2986bfa176cb713fad8b0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-19 07:03:23.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a0d358a48fb2da55f1a56f9ad43835946e183322","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 10:15:35.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ed276a4a92ed7814cd14262274b546bcf068bba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 10:16:24.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"9fbf33f24ca0c826e63a85b4fb6fb50677e1a61e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:37:42.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43df296c4fa4b48aca27f8c6583f73f\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe70ad9fa5b54e40bff189fa1f388580 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcedf18864d84c638cbfb26ea989f56a : SUCCESS in 1m 48s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5be87d3d674144689d71fbf238e206c4 : SUCCESS in 12m 43s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/018b6d4d5a1e4959b79e2f5ddeefb64a : SUCCESS in 8m 21s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/83ac3931383645429d84319e8efb31f7 : SUCCESS in 55m 54s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cfd76374edb4e7f93f064cb599b5458 : SUCCESS in 1h 14m 47s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/792d6da3ba14497bb5fa9b527d0bb940 : SUCCESS in 54m 27s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb7ef10788cd450c9581b6fed6dee43f : SUCCESS in 1h 03m 01s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fae08ad05b004d7daee328dc05bf44e4 : SUCCESS in 28m 56s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/94e621d426ed45ff9714614d4e43ada3 : SUCCESS in 18m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e81c3fc26f0454ca625b44eb58c6217 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":8},{"id":"de40df12090445a46173175952e33ef91945305e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:37:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"26eaf55c32b4114dde5dd76d742fa6e4d68919d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:38:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f9938d9a874d969336a11a4aaeb986\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9249518b00704c69a863b759409e5d57 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"54677589b12eca8d92a169490b0163f35fd4d0b9","revisions":{"3fcf4464b4cb73a2616c59a016f54379ea010a82":{"kind":"REWORK","_number":1,"created":"2025-08-18 01:11:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/1"}}},"commit":{"parents":[{"commit":"21fc8f0620c4f2926ec81fa7d4618764ec810a87","subject":"Remove unused variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/21fc8f0620c4f2926ec81fa7d4618764ec810a87"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:11:32.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3fcf4464b4cb73a2616c59a016f54379ea010a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3fcf4464b4cb73a2616c59a016f54379ea010a82"}]},"branch":"refs/heads/master"},"d90b3f88b3039ea4e6b3f69f09f7dfe90cdea59e":{"kind":"REWORK","_number":2,"created":"2025-08-18 01:44:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/2"}}},"commit":{"parents":[{"commit":"21fc8f0620c4f2926ec81fa7d4618764ec810a87","subject":"Remove unused variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/21fc8f0620c4f2926ec81fa7d4618764ec810a87"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:44:33.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d90b3f88b3039ea4e6b3f69f09f7dfe90cdea59e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d90b3f88b3039ea4e6b3f69f09f7dfe90cdea59e"}]},"branch":"refs/heads/master"},"789c24a1c4f39c9e604e9995a949fda3d482a9ad":{"kind":"REWORK","_number":3,"created":"2025-08-18 01:47:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/3"}}},"commit":{"parents":[{"commit":"617afad85bfe0a2f63160cfa4a5b6463c4aa557f","subject":"Remove redundant variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/617afad85bfe0a2f63160cfa4a5b6463c4aa557f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:47:09.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/789c24a1c4f39c9e604e9995a949fda3d482a9ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/789c24a1c4f39c9e604e9995a949fda3d482a9ad"}]},"branch":"refs/heads/master"},"3a38c458f645646a341094bc87ffc280aecb806e":{"kind":"REWORK","_number":4,"created":"2025-08-18 02:34:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/4"}}},"commit":{"parents":[{"commit":"617afad85bfe0a2f63160cfa4a5b6463c4aa557f","subject":"Remove redundant variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/617afad85bfe0a2f63160cfa4a5b6463c4aa557f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 02:34:34.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3a38c458f645646a341094bc87ffc280aecb806e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3a38c458f645646a341094bc87ffc280aecb806e"}]},"branch":"refs/heads/master"},"9d4632e491ff10b86401d0f927c9b15901e573d0":{"kind":"REWORK","_number":5,"created":"2025-08-18 08:53:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/5"}}},"commit":{"parents":[{"commit":"617afad85bfe0a2f63160cfa4a5b6463c4aa557f","subject":"Remove redundant variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/617afad85bfe0a2f63160cfa4a5b6463c4aa557f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 08:52:27.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9d4632e491ff10b86401d0f927c9b15901e573d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9d4632e491ff10b86401d0f927c9b15901e573d0"}]},"branch":"refs/heads/master"},"5f4832a81ae38b558b3abb7c199eddaf41ead593":{"kind":"REWORK","_number":6,"created":"2025-08-18 11:19:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/6"}}},"commit":{"parents":[{"commit":"61fccb735e0763e9d4342cab4351129e028ea838","subject":"Remove redundant variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/61fccb735e0763e9d4342cab4351129e028ea838"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 11:18:54.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5f4832a81ae38b558b3abb7c199eddaf41ead593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5f4832a81ae38b558b3abb7c199eddaf41ead593"}]},"branch":"refs/heads/master"},"2db6aa0d3a6a7fd100016dd2f0b53112249c1768":{"kind":"REWORK","_number":7,"created":"2025-08-18 11:23:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/7"}}},"commit":{"parents":[{"commit":"61fccb735e0763e9d4342cab4351129e028ea838","subject":"Remove redundant variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/61fccb735e0763e9d4342cab4351129e028ea838"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 11:23:16.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2db6aa0d3a6a7fd100016dd2f0b53112249c1768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2db6aa0d3a6a7fd100016dd2f0b53112249c1768"}]},"branch":"refs/heads/master"},"54677589b12eca8d92a169490b0163f35fd4d0b9":{"kind":"REWORK","_number":8,"created":"2025-08-18 14:20:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/957636/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/957636/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/957636/8"}}},"commit":{"parents":[{"commit":"61fccb735e0763e9d4342cab4351129e028ea838","subject":"Remove redundant variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/61fccb735e0763e9d4342cab4351129e028ea838"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 01:06:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 14:18:05.000000000","tz":540},"subject":"Remove support for standalone neutron-server","message":"Remove support for standalone neutron-server\n\nIt was removed from neutron during this cycle due to removal of\nthe eventlet library.\n\nDepends-on: https://review.opendev.org/952659\nDepends-on: https://review.opendev.org/957585\nChange-Id: Ibf260264d83f285008155223ada30423dc26254e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54677589b12eca8d92a169490b0163f35fd4d0b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54677589b12eca8d92a169490b0163f35fd4d0b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
