)]}'
{"id":"openstack%2Fpuppet-neutron~963365","triplet_id":"openstack%2Fpuppet-neutron~master~I86aae720b1cc728262bef4af2adaa2168777b2be","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-09 12:38:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I86aae720b1cc728262bef4af2adaa2168777b2be","subject":"dnm: Testing","status":"ABANDONED","created":"2025-10-08 10:59:55.000000000","updated":"2025-10-09 12:38:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"69d6be876e39ce67d8faedcc087d52906a73e0c9","_number":963365,"virtual_id_number":963365,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-09 02:58:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-08 11:20:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e61f6c30cb7643ff9e8bf56478c93987d7bc513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 10:59:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b111c3fb74de7f399b878341816356d574cc1322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 11:20:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6bdfdadf48d466eb68dacd4a91ca48c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95f086c4b05e440eb13d9f4f466d3f5e : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39e9fcb398d74c9580a520652e38fa02 : SUCCESS in 2m 53s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f70887e0e64480f8cc7641ddce5e9d3 : FAILURE in 7m 35s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a460d79fba8e4a22a7939675b2e449bd : FAILURE in 3m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/187ec094dfbb4f0dacaa7d6f190a983a : FAILURE in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c1f5e775fb6edd63b421cd836155682c752faf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 12:48:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"12640913ccaba43ece0afa07bb133dec72f50e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 13:11:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59edfa80eb1b4f1a917e80488a0578a6\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1327363941e44da58b6646a081558383 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/456ec22fe78f492d89345da1db9999e8 : SUCCESS in 2m 52s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9089e18af976457fa1f33e95f527a706 : SUCCESS in 15m 34s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a2f073e79ff428083ab46aee1aa1f4c : SUCCESS in 13m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f66f969350a849d5b0010433b467fbf5 : SUCCESS in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0c1dd387f08654cfc3b5e7f3cab4c621dc428d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 13:42:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6921ab1c83773645e0bcc30c6936207d0210110b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 13:42:12.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"735ae013adec7300158c11751b6929a5a2f0fb9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 14:01:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc3b66f56ed46bbb14d84f417a98031\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/659c5e67c168495bbe0f5900930876bf : SUCCESS in 1m 54s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b9ed125ca2e4db18dfc652124e733c4 : SUCCESS in 3m 09s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ab2311daaf54531a5dd019714b3c027 : SUCCESS in 12m 38s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c383210ccca44a9ca5d0550cf3fb20f4 : FAILURE in 12m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e1496a84c6b427c8a4303836a9d4392 : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52bcd542d56f48b9e05bc63f10d689abe98e436b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-09 02:29:57.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2f8671e459fa11d1da3bb25a66df4141c788c0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 02:58:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2101fb8469a0404b999992a02af4c9b3\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a4e9ced3f6c418f88b0c107fb6b3692 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68deb955f67149e4b6b30335a72965a4 : SUCCESS in 2m 50s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/01f5e30268e34706afed305e6ee8cc47 : SUCCESS in 22m 04s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f469397ae49547409d38374d7bebfbaf : SUCCESS in 13m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2723373041474399abd837c0f4a83943 : SUCCESS in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"69d6be876e39ce67d8faedcc087d52906a73e0c9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-09 12:38:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"93e0f33d885e0f2bda999fc5b267bd82dbb5793a","revisions":{"9fb5bbd5c45f02752268017d6f9ff33f96e24b61":{"kind":"REWORK","_number":1,"created":"2025-10-08 10:59:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/963365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/963365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/1"}}},"commit":{"parents":[{"commit":"f3bc3e95c4c3f424e7fc06271f3651a4e5db164d","subject":"neutron_router: Allow absent fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f3bc3e95c4c3f424e7fc06271f3651a4e5db164d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 10:59:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 10:59:46.000000000","tz":540},"subject":"dnm: Testing","message":"dnm: Testing\n\nChange-Id: I86aae720b1cc728262bef4af2adaa2168777b2be\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9fb5bbd5c45f02752268017d6f9ff33f96e24b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9fb5bbd5c45f02752268017d6f9ff33f96e24b61"}]},"branch":"refs/heads/master"},"a8d0a1d8de96eeb82a47d878673ceddb6ae7a3c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-08 12:48:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/963365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/963365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/2"}}},"commit":{"parents":[{"commit":"635ad914a22cffb062a91854f8ae8cc0008db34e","subject":"neutron_router: Allow absent fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/635ad914a22cffb062a91854f8ae8cc0008db34e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 10:59:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 12:48:08.000000000","tz":540},"subject":"dnm: Testing","message":"dnm: Testing\n\nChange-Id: I86aae720b1cc728262bef4af2adaa2168777b2be\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a8d0a1d8de96eeb82a47d878673ceddb6ae7a3c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a8d0a1d8de96eeb82a47d878673ceddb6ae7a3c9"}]},"branch":"refs/heads/master"},"7b066bbf61f77008633a8772845f6f2f9d0f78e1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-08 13:42:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/963365/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/963365/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/3"}}},"commit":{"parents":[{"commit":"c00c7ea8ca945e86d0b6b4966934ad32de185eb1","subject":"Remove packaging files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c00c7ea8ca945e86d0b6b4966934ad32de185eb1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 10:59:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 13:42:06.000000000","tz":0},"subject":"dnm: Testing","message":"dnm: Testing\n\nChange-Id: I86aae720b1cc728262bef4af2adaa2168777b2be\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7b066bbf61f77008633a8772845f6f2f9d0f78e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7b066bbf61f77008633a8772845f6f2f9d0f78e1"}]},"branch":"refs/heads/master","description":"Rebase"},"93e0f33d885e0f2bda999fc5b267bd82dbb5793a":{"kind":"REWORK","_number":4,"created":"2025-10-09 02:29:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/963365/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/963365/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/963365/4"}}},"commit":{"parents":[{"commit":"c00c7ea8ca945e86d0b6b4966934ad32de185eb1","subject":"Remove packaging files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c00c7ea8ca945e86d0b6b4966934ad32de185eb1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 10:59:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-09 02:29:57.000000000","tz":0},"subject":"dnm: Testing","message":"dnm: Testing\n\nChange-Id: I86aae720b1cc728262bef4af2adaa2168777b2be\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/93e0f33d885e0f2bda999fc5b267bd82dbb5793a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/93e0f33d885e0f2bda999fc5b267bd82dbb5793a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
