)]}'
{"id":"openstack%2Fpuppet-neutron~963469","triplet_id":"openstack%2Fpuppet-neutron~master~Ia11cc81c9a2cd182978cd6b666793120f531614a","project":"openstack/puppet-neutron","branch":"master","topic":"bug/2127133","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2025-10-09 09:54:58.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-09 11:25:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia11cc81c9a2cd182978cd6b666793120f531614a","subject":"DNM - neutron_router nil:NilClass","status":"ABANDONED","created":"2025-10-08 22:10:12.000000000","updated":"2025-10-09 11:25:42.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"465b5bbb09bb9312d8c74c2c73ababb250e54312","_number":963469,"virtual_id_number":963469,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-09 00:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 02:31:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"5911e8f5f3a0bc256da09a75a57a2798031f61f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 22:10:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beb92a08269ee07be2d795e4fcb183d1c18ca369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 22:44:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be5bd99f2558c964c2ff7b1d51fe7f57784d7187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 23:24:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"187ba31d4b59cdbaf002f4608a6dc0fed2a7787e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 00:53:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a16759c25eaf44c898770549884a0fec\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29e508fef9ea43a8a10d29c4697b3314 : SUCCESS in 1m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07f6090fb17c4a689724ca7e4107720b : SUCCESS in 2m 51s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/2e22065c9a95487bb8fbcb27528c5b31 : FAILURE in 5m 35s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa462790aa084cc3a177130c43fb8d48 : FAILURE in 4m 46s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/609f921a0eb44db29fe89bd7a4efe469 : FAILURE in 11m 20s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a60bc78e88743878b9b3fb8a9f0ef8d : SUCCESS in 55m 27s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e127a0442953439f82915f0cbbf88b43 : SUCCESS in 1h 06m 55s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25e4db358b204cddb46c5a33b8c1911f : FAILURE in 35m 25s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/228e0c0ecdc24841a690dd773de4e60b : SUCCESS in 45m 55s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/011905d9ecef4040bcb09655c1cef42c : FAILURE in 51m 38s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05041c55c29e4e10afe92c7a60ff6e09 : SUCCESS in 1h 06m 01s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5e8c8473c80446b8ca3b109d390ec65 : SUCCESS in 1h 22m 37s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d0045e0817542e4ad5e1b8e9fdd339b : SUCCESS in 1h 00m 48s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/beb0a3bbbe824710862331ee94b8228b : SUCCESS in 1h 00m 17s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2703509cc7d4dff958a9a838b94b6dc : SUCCESS in 1h 02m 32s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee5de1dcb0e347b781ea858b4846bd60 : SUCCESS in 1h 13m 25s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/975b436abf6e40029835c4febe7aeaa1 : SUCCESS in 1h 20m 28s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa9dd0a0f5324bd0859bd56f6aa22197 : SUCCESS in 48m 29s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a0ec36ecb6947b8a82f49630f8ec0a9 : SUCCESS in 56m 33s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcca312536a749a09752f3cdc77de01d : SUCCESS in 56m 33s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1da802326f5c42eeb5fd3b5a902d5501 : SUCCESS in 11m 15s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ac4672f9e4f48168f55fc3dcb284e60 : SUCCESS in 11m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b08e9102a1b64b28ada4b182f24c7078 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8995b208a0f8b3e6d649eb93d1608a7b409ff71c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-09 02:31:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eb380a912928778ba85bd73cb04ed275a3b673d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-09 09:54:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"1007a78985e663c61adbbc23b9a7db389f7fc9e0","tag":"autogenerated:gerrit:restore","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-09 09:58:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"2064b25d740405efafb4c33dc730b1a74b04a734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-09 10:21:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"47c01a609dafb4c3b65cf73744624a9288f5c8d3","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-09 10:48:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"128166f6401dd98126856de58b033080d6b1eefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-09 10:48:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f2806eff144b23acb2313ca62305467b6f1f2edd","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-09 11:25:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"465b5bbb09bb9312d8c74c2c73ababb250e54312","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-09 11:25:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"736455cbcc021cce5f42da652bc64486f92db649","revisions":{"0e1c243918b127e6941ddb6c0d8977323219f4ef":{"kind":"REWORK","_number":1,"created":"2025-10-08 22:10:12.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/69/963469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/963469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/1"}}},"commit":{"parents":[{"commit":"c00c7ea8ca945e86d0b6b4966934ad32de185eb1","subject":"Remove packaging files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c00c7ea8ca945e86d0b6b4966934ad32de185eb1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:09:06.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:09:56.000000000","tz":120},"subject":"WIP - neutron_router handle OVN nil values","message":"WIP - neutron_router handle OVN nil values\n\nChange-Id: Ia11cc81c9a2cd182978cd6b666793120f531614a\nCloses-Bug: LP#2127133\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0e1c243918b127e6941ddb6c0d8977323219f4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0e1c243918b127e6941ddb6c0d8977323219f4ef"}]},"branch":"refs/heads/master"},"5c465d1fa0f5502b44848b783b07919beebc58c7":{"kind":"REWORK","_number":2,"created":"2025-10-08 22:44:07.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/69/963469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/963469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/2"}}},"commit":{"parents":[{"commit":"c00c7ea8ca945e86d0b6b4966934ad32de185eb1","subject":"Remove packaging files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c00c7ea8ca945e86d0b6b4966934ad32de185eb1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:09:06.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:43:14.000000000","tz":120},"subject":"WIP - neutron_router nil:NilClass","message":"WIP - neutron_router nil:NilClass\n\nChange-Id: Ia11cc81c9a2cd182978cd6b666793120f531614a\nCloses-Bug: LP#2127133\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5c465d1fa0f5502b44848b783b07919beebc58c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5c465d1fa0f5502b44848b783b07919beebc58c7"}]},"branch":"refs/heads/master"},"091148abad8861654ed0dcced9f25c9891535c81":{"kind":"REWORK","_number":3,"created":"2025-10-08 23:24:23.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/69/963469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/963469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/3"}}},"commit":{"parents":[{"commit":"c00c7ea8ca945e86d0b6b4966934ad32de185eb1","subject":"Remove packaging files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c00c7ea8ca945e86d0b6b4966934ad32de185eb1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:09:06.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 23:24:17.000000000","tz":120},"subject":"WIP - neutron_router nil:NilClass","message":"WIP - neutron_router nil:NilClass\n\nChange-Id: Ia11cc81c9a2cd182978cd6b666793120f531614a\nCloses-Bug: LP#2127133\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/091148abad8861654ed0dcced9f25c9891535c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/091148abad8861654ed0dcced9f25c9891535c81"}]},"branch":"refs/heads/master"},"0058b43606582622e6e35156803a4f5204c109cf":{"kind":"REWORK","_number":4,"created":"2025-10-09 10:21:48.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/69/963469/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/963469/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/4"}}},"commit":{"parents":[{"commit":"c00c7ea8ca945e86d0b6b4966934ad32de185eb1","subject":"Remove packaging files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c00c7ea8ca945e86d0b6b4966934ad32de185eb1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:09:06.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-09 10:20:27.000000000","tz":120},"subject":"DNM - neutron_router nil:NilClass","message":"DNM - neutron_router nil:NilClass\n\nSlight variation on change[1] for testing purposes.\n\n[1] https://review.opendev.org/c/openstack/puppet-neutron/+/963364\n\nChange-Id: Ia11cc81c9a2cd182978cd6b666793120f531614a\nRelated-Bug: LP#2127133\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0058b43606582622e6e35156803a4f5204c109cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0058b43606582622e6e35156803a4f5204c109cf"}]},"branch":"refs/heads/master"},"736455cbcc021cce5f42da652bc64486f92db649":{"kind":"REWORK","_number":5,"created":"2025-10-09 10:48:43.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/69/963469/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/963469/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/963469/5"}}},"commit":{"parents":[{"commit":"c00c7ea8ca945e86d0b6b4966934ad32de185eb1","subject":"Remove packaging files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c00c7ea8ca945e86d0b6b4966934ad32de185eb1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:09:06.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-09 10:48:34.000000000","tz":120},"subject":"DNM - neutron_router nil:NilClass","message":"DNM - neutron_router nil:NilClass\n\nSlight variation on change[1] for testing purposes.\n\n[1] https://review.opendev.org/c/openstack/puppet-neutron/+/963364\n\nChange-Id: Ia11cc81c9a2cd182978cd6b666793120f531614a\nRelated-Bug: LP#2127133\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/736455cbcc021cce5f42da652bc64486f92db649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/736455cbcc021cce5f42da652bc64486f92db649"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
