)]}'
{"id":"openstack%2Fpuppet-neutron~978000","triplet_id":"openstack%2Fpuppet-neutron~master~I38dd7d3eb4c849273601b1f716d04a8fb556af5e","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-26 23:31:20.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2026-02-26 22:23:04.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I38dd7d3eb4c849273601b1f716d04a8fb556af5e","subject":"Deprecate support for [ml2] tenant_network_types","status":"MERGED","created":"2026-02-25 17:18:52.000000000","updated":"2026-02-26 23:32:14.000000000","submitted":"2026-02-26 23:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978000","meta_rev_id":"aaa7d44716929b85cf0cc7c08c9c60bf9f70163d","_number":978000,"virtual_id_number":978000,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-26 23:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2026-02-26 07:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-02-26 22:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-26 10:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2026-02-26 07:52:22.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2026-02-26 22:23:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-25 17:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 07:51:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-02-26 10:38:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-26 10:38:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"70db996b1169e56b46d71446d942c8385b02fb90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-25 17:18:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1621a01ae95f5492880ccb4cce46e1b66302cf8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-25 17:19:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8403de7f8119bf74c1431ce2d5de3124c756e1c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 17:40:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90273bd15b3e4000b474c0325b78bf6b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ead3d4a5b9df48e1a84063e6839898b8 : FAILURE in 3m 19s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/123ba8902ce44917ac059807c5fa74b9 : FAILURE in 2m 55s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/9fe9bd3b61cc4bec8ad02befcdb23aba : FAILURE in 3m 38s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a66547c7e88f442a9948916dc1036d55 : FAILURE in 3m 19s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25a6fbc1c83c40669d0161a610d3c2ef : FAILURE in 4m 32s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c8bdacd8de44671877ea1562a67d6a7 : FAILURE in 5m 35s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b84a07c82c14832a3d3d3c256ed39bb : FAILURE in 5m 34s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5af4d0e6a1bd442db5be288a241447ea : FAILURE in 3m 40s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/397fa90951e34d70801f3144b6e891e6 : FAILURE in 4m 40s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a7bc685d234467ea1f0073d96ea38bb : FAILURE in 5m 59s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7807f6d28a9844949709cfd3d9f975d5 : FAILURE in 4m 53s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a68563c85e08469ebe2ec7094b69f8c5 : FAILURE in 10m 10s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/41db5b342cd143e3a78dac1b2e01eb7c : FAILURE in 9m 36s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa71a0ed8687467a9a1853678ed78d31 : FAILURE in 4m 54s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/12b525b2cf8a40268fb56aa11fab9157 : FAILURE in 7m 07s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3eafc78b5504f249b1a38a6a75d53fe : FAILURE in 8m 43s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfa947b1d75f4e83a22f339bb827a50d : FAILURE in 8m 28s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ae720eca6a843a883f7f42efb6eece9 : FAILURE in 9m 40s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/026c020dd92642e387394d060dbec9f5 : FAILURE in 9m 05s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97eda7cab01c4a21a162eaba55c1c380 : FAILURE in 9m 10s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/38bb706d78f84cfb816c03c4b8df0986 : FAILURE in 7m 13s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f87cbc73ca9f4850be758a2a001f673e : FAILURE in 4m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e04f3f9e7204083b3932e584721b2be : FAILURE in 5m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7ac6f06a365471080d666a9ce805983 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"361f649fdbf1b4449add5d082e7c79653decdd55","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-25 23:12:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b534143de8a3949f7b2e6cd78cc32bc6d90d624f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-26 00:24:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8079c70ab6570eb5517838d60676c1d51e2aec42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 02:03:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90143f2b06bb4864ab67e625ecf056ae\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7af224c876ff4e26ae567d9f7aa96440 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/806847e5e4204ee990ad44504304152c : SUCCESS in 1m 50s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/395cb14b68774e08be2a44648eae3f40 : SUCCESS in 6m 03s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a562fc91d2e4ed0b29f86d09a8a86d2 : SUCCESS in 11m 29s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15c53412a4654730981340fbff3ee5b3 : SUCCESS in 11m 16s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfba1ebfa8e44d109806ee9f902e44be : SUCCESS in 46m 44s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56887f981fd346558cbe3bfe056a0732 : FAILURE in 1h 11m 05s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0315cf51101e401eb6a922838bce4eb6 : FAILURE in 1h 06m 44s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d3dcbe616be4f8b94b4423a8926138d : FAILURE in 57m 25s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a851002231948cc8db94aba69217d02 : FAILURE in 1h 00m 06s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61b97855568f407685bfb8ddc340f7a5 : SUCCESS in 32m 01s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd9c7e9ceba749418fef839db93acd52 : SUCCESS in 1h 23m 24s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed37aaac11764d38a9396ba5e8ded656 : SUCCESS in 58m 10s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2fe2a6333584ffea9841709d4d59242 : SUCCESS in 58m 33s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/756321c0c8f54f88872721cbe4418ed2 : SUCCESS in 42m 52s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abcf1a3c662644eeab9737e3d0d259ea : SUCCESS in 1h 06m 42s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f72c548e385742099213d492cb360266 : SUCCESS in 1h 28m 13s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2947492bf46d42bea854b51c54de47f5 : SUCCESS in 57m 41s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/12a0b9347a48491db1aeb78410931f8f : SUCCESS in 54m 50s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2cd69d1f31c4de1bef666bf02cb8ae0 : SUCCESS in 1h 03m 27s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2631d10db6a4477af4a736bddfa16b4 : SUCCESS in 20m 00s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9140f078a5404d7299938282877c6b18 : SUCCESS in 12m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1efbda1561534100a8c2aa4caa0da685 : SUCCESS in 13m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd690227817c40aaab4ab685726603ba : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3},{"id":"17216e4cf38b28c10172d1c65e6d5f05aa3d9cf3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-26 07:51:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"99e0acb44ed506487bdc506f62ed44cb956cc080","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-26 07:51:59.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"c77977fe4232d0409086efe6cfda7694e6573e6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 07:52:01.000000000","message":"Patch Set 3: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":3},{"id":"4ac189a3c29289ee6c1dcac66b73747ca466029c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-26 07:52:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5696349ed81a540760c900d3bb010de5892e425f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 09:19:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3494bcf6ceac4ff4a7783cff6a74497e\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afa08ee1f7d84e0abfd1eb71c46d3784 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6899223647e49f7a4910b5e1e55001e : SUCCESS in 2m 10s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/b993c5f3d7c44d0192f150fbf8f18f2d : SUCCESS in 7m 09s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/450b4db9bb78499fa83683a17b941803 : SUCCESS in 12m 16s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32b97e65bed54960a6dff738f810914d : SUCCESS in 11m 26s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea952408193b4763be8e63a006c3e32d : SUCCESS in 1h 01m 17s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9ee3c3a5f23404e9dfd33b983a96fa8 : FAILURE in 1h 20m 50s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98aebe84a30244a4bb317c704ed00f2b : FAILURE in 1h 12m 47s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ada9a5705ff94e3688aa855fdc208859 : FAILURE in 1h 10m 50s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e426deef13af4104bb5bc2a83c330b8b : FAILURE in 1h 08m 22s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5c5d18732644cb9a84fcbd58f7f9197 : SUCCESS in 1h 07m 01s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1a212ed24a742ebb511e5fbede2242d : SUCCESS in 1h 21m 26s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f8cd690ecd04aba884fd4821f7a6acf : SUCCESS in 1h 05m 24s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4eb2f19d61645dc85db1a94282dd3a9 : SUCCESS in 53m 24s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8068fc51a0b34c67a017e9b0ab42e5be : SUCCESS in 56m 40s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7344aef165324316afb1999498b1fdcc : SUCCESS in 1h 08m 33s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e281af1de99443fa64f49112a33437a : SUCCESS in 1h 17m 07s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3f63bb5f97c4747a2887b104a60236d : SUCCESS in 49m 42s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc5d3c3fd9e5472faad5ccad450ea7b7 : SUCCESS in 51m 31s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b66d69e3853e48608b20989e7b6d3c1f : SUCCESS in 29m 04s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b24f437fc6e0446ca5761de97d67947d : SUCCESS in 21m 58s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fde90d808f834e1eba8c6b96823383e6 : SUCCESS in 17m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f9a1c6d19c74ecaad6d7a6cbd45d54d : SUCCESS in 17m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f232da5f79a4497baecdfdc5a0e5a3fb : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"4495dc7c80af2ca59db9e44dd08e7b7a99c43a4d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 10:38:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a13791adb2ca2ea3adecc8f43d5467b53b171f15","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-02-26 22:23:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8a393349f57021ff403fdcbbd432d1f4574d683b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 22:23:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"03c7859933bab877d603707ed7e3000f2b1ab3e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 23:31:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e8b0e52b02142c789f947e345f4fe26\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98a8c8daf50042f8bdad3c5c1ba129f4 : SUCCESS in 2m 22s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c40b937e5f140aaad4578a082576048 : SUCCESS in 1m 51s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/50fe35992cb54534938a1d6396e9c584 : SUCCESS in 7m 57s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df0a3c3340944eceaa7d7889561303d3 : SUCCESS in 11m 39s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b45d76a981e4809aa009b550ecd6128 : SUCCESS in 56m 17s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28f479d7b5a4415a9243c5be187e497a : SUCCESS in 37m 30s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9abbe3ca1d3b4773acccd8b8617419bf : SUCCESS in 1h 02m 03s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dc581c7db8f4da5baa5dedeca310d70 : SUCCESS in 57m 54s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a62bfeab86554952b6c91dba6073e9ba : SUCCESS in 35m 42s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/392e59822e9c4819af81101b591e159b : SUCCESS in 19m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce888e807cf54c2b908fdef78b9b12de : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"382bbcc022ffe95ba1a63f28638491b1b0cb5ffc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 23:31:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"aaa7d44716929b85cf0cc7c08c9c60bf9f70163d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 23:32:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7a18582d304edfa0bc7d9effe18450\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c55c82bd957427aaf7b2dcaeed894ec : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3fc75765a4c9f31839841952ef965d4ec25f50be","revisions":{"e40d56543f3f4c45a7c8900fa0672cc4736747ce":{"kind":"REWORK","_number":1,"created":"2026-02-25 17:18:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/978000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/978000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/1"}}},"commit":{"parents":[{"commit":"029d1e65db46d716a20b77bb7db3bcc43e85c55a","subject":"Merge \"Remove support for [ovn] ovn_emit_need_to_frag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/029d1e65db46d716a20b77bb7db3bcc43e85c55a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-25 17:08:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-25 17:18:19.000000000","tz":540},"subject":"Deprecate support for [ml2] tenant_network_types","message":"Deprecate support for [ml2] tenant_network_types\n\nThe option was deprecated in neutron due to renaming. The old parameter\ntriggers configuration of tenant_network_types to keep our testing\npassing with older release of OpenStack.\n\nAlso replace references to \u0027tenant\u0027.\n\nDepends-On: https://review.opendev.org/972591\nChange-Id: I38dd7d3eb4c849273601b1f716d04a8fb556af5e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e40d56543f3f4c45a7c8900fa0672cc4736747ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e40d56543f3f4c45a7c8900fa0672cc4736747ce"}]},"branch":"refs/heads/master"},"f87ff9c7b22a002c810c80918219b86ed0437303":{"kind":"REWORK","_number":2,"created":"2026-02-25 17:19:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/978000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/978000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/2"}}},"commit":{"parents":[{"commit":"029d1e65db46d716a20b77bb7db3bcc43e85c55a","subject":"Merge \"Remove support for [ovn] ovn_emit_need_to_frag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/029d1e65db46d716a20b77bb7db3bcc43e85c55a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-25 17:08:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-25 17:19:42.000000000","tz":540},"subject":"Deprecate support for [ml2] tenant_network_types","message":"Deprecate support for [ml2] tenant_network_types\n\nThe option was deprecated in neutron due to renaming. The old parameter\ntriggers configuration of tenant_network_types to keep our testing\npassing with older release of OpenStack.\n\nAlso replace references to \u0027tenant\u0027.\n\nDepends-On: https://review.opendev.org/972591\nChange-Id: I38dd7d3eb4c849273601b1f716d04a8fb556af5e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f87ff9c7b22a002c810c80918219b86ed0437303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f87ff9c7b22a002c810c80918219b86ed0437303"}]},"branch":"refs/heads/master"},"3fc75765a4c9f31839841952ef965d4ec25f50be":{"kind":"REWORK","_number":3,"created":"2026-02-26 00:24:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/978000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/978000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/978000/3"}}},"commit":{"parents":[{"commit":"029d1e65db46d716a20b77bb7db3bcc43e85c55a","subject":"Merge \"Remove support for [ovn] ovn_emit_need_to_frag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/029d1e65db46d716a20b77bb7db3bcc43e85c55a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-25 17:08:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-26 00:24:44.000000000","tz":540},"subject":"Deprecate support for [ml2] tenant_network_types","message":"Deprecate support for [ml2] tenant_network_types\n\nThe option was deprecated in neutron due to renaming. The old parameter\ntriggers configuration of tenant_network_types to keep our testing\npassing with older release of OpenStack.\n\nAlso replace references to \u0027tenant\u0027.\n\nDepends-On: https://review.opendev.org/972591\nChange-Id: I38dd7d3eb4c849273601b1f716d04a8fb556af5e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3fc75765a4c9f31839841952ef965d4ec25f50be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3fc75765a4c9f31839841952ef965d4ec25f50be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
