)]}'
{"id":"openstack%2Fpuppet-neutron~980490","triplet_id":"openstack%2Fpuppet-neutron~master~Idfe6a178f17e47643c5e7458df7404cbf4c14392","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-16 23:40:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idfe6a178f17e47643c5e7458df7404cbf4c14392","subject":"Add support for [DEFAULT] dnsmasq_txt_record","status":"MERGED","created":"2026-03-13 16:07:18.000000000","updated":"2026-03-16 23:41:05.000000000","submitted":"2026-03-16 23:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980490","meta_rev_id":"5ac82c53bbd328a27b3214e358b93ae1dd193683","_number":980490,"virtual_id_number":980490,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-16 23:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2026-03-16 14:13:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2026-03-16 22:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2026-03-16 22:06:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 18:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-16 14:13:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2026-03-16 22:06:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"58bd8c3459be4ac0db462e972d92ecabe1e537bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-13 16:07:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6444109eaa368067670c132df85ef404c188582f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-13 16:08:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d23ab38e5a2669e20692efcc39c9b17a239bdf5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 18:01:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60dba38382c441ebd170a8b9e68bbdb\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b4b0d62e76744a881c06bb849bda141 : SUCCESS in 2m 12s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4370027b91343c88f46b17d7c9475a0 : SUCCESS in 3m 23s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e05be012a8264fc99f032a793d8460df : SUCCESS in 9m 42s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c55ec55e912e4ebc83896df9d54e9f56 : SUCCESS in 9m 03s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26107a6e4add481d909847087576a70b : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39670c35d94644a0833d5d900acc7e9f : SUCCESS in 58m 28s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec69a4cb994b4972b6b6e0bddc479b8c : FAILURE in 1h 43m 22s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/323aba76c0bd4abf8d8b8b1002a242d9 : FAILURE in 1h 01m 29s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20bb528404a2464890109b5670a33c48 : FAILURE in 1h 13m 56s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a702aab63b8b4f98832cf78a09d6db82 : FAILURE in 1h 06m 54s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf67dd4010a54abcba61ef2b9a759204 : SUCCESS in 1h 06m 54s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40b8da39cb2c445983174454b85a88f4 : SUCCESS in 45m 21s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d83be090276540d499f300ec42beb7ef : SUCCESS in 1h 00m 27s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c19760fa29446db964ba574b837f29a : SUCCESS in 55m 54s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/90baf80cfab74e768bd51ffde8e7e30e : SUCCESS in 52m 49s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a378698e5804996a9c3c2042d0a3bfa : SUCCESS in 1h 13m 15s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5bfb6a1818243d09c31a969e0880a8e : SUCCESS in 1h 18m 03s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/333c8eaef3494e7ca22dbc0ae7e3e108 : SUCCESS in 59m 21s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/83af4f0a807e4da49726690a983c50f6 : SUCCESS in 1h 05m 25s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0237128780c849498a87691ef8f4e733 : SUCCESS in 53m 07s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/19aeb6aef31b46be88ce5172f02b0ac3 : SUCCESS in 18m 12s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f60c3eb80fa1492c9235c25d937eecbb : SUCCESS in 13m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d783f580db246f0a8308cd9e7520122 : SUCCESS in 13m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63372102e0d84773ae6b12ec35881f2f : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a4989feb61ff551cb65bf3f9e529a2bd8a76c530","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-03-16 14:13:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"31fff73eaae91d441194f1e10ae569f9709d6e51","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-03-16 22:06:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c34907243af9cc464c2ca1992a752f8a4aa9af5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 22:06:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"59ad7ca57ecd7ed1aeda300a838d0bdc79c39946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:40:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fd2c0561cd44be4b9a2ed577f2d7068\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af73309d81b141728c749ac75ecddbbd : SUCCESS in 3m 29s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23dbdc75651240daa766e3fe47d42555 : SUCCESS in 2m 55s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/7e9fdab0cfa64dbf969625288744ec8d : SUCCESS in 9m 25s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbc694d1bab84fa181f0eedd76c62130 : SUCCESS in 5m 30s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c49415b98c74d0d8ace8e2a105a69d2 : SUCCESS in 56m 16s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1b3b6e122c24dce98edf8d2e77b5d28 : SUCCESS in 1h 27m 10s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b93a25a5480403bb1140f74b469b66d : SUCCESS in 30m 41s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f3cf4470c8c49bcb2b3438bb239143f : SUCCESS in 31m 37s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f28505007faf4216b1972fe5f059843b : SUCCESS in 56m 21s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb9878fb444a41acaf466bca8e2bf17b : SUCCESS in 22m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/804eb35d7a52402888157299d03798ff : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c29be1c00aa7c40d8d8cec5daeb21c5755e59f9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:40:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5ac82c53bbd328a27b3214e358b93ae1dd193683","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:41:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fe32dba837e40728ee67262b1ef1fcb\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62a3c47931e24e24af2f820a27ec2eae : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7923d17c0ff7d7d6de2939bf3072c03766f6097","revisions":{"b34f11634c31daeacb2421597631e993ba7bf8a0":{"kind":"REWORK","_number":1,"created":"2026-03-13 16:07:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/980490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/90/980490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/1"}}},"commit":{"parents":[{"commit":"3fc75765a4c9f31839841952ef965d4ec25f50be","subject":"Deprecate support for [ml2] tenant_network_types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3fc75765a4c9f31839841952ef965d4ec25f50be"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-13 16:05:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-13 16:06:58.000000000","tz":540},"subject":"Add support for [DEFAULT] dnsmasq_txt_record","message":"Add support for [DEFAULT] dnsmasq_txt_record\n\nDepeneds-on: https://review.opendev.org/950486\nChange-Id: Idfe6a178f17e47643c5e7458df7404cbf4c14392\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b34f11634c31daeacb2421597631e993ba7bf8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b34f11634c31daeacb2421597631e993ba7bf8a0"}]},"branch":"refs/heads/master"},"d7923d17c0ff7d7d6de2939bf3072c03766f6097":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-13 16:08:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/980490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/90/980490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/90/980490/2"}}},"commit":{"parents":[{"commit":"eaf3669186fd36162bcb8f95883603e1e7041128","subject":"Merge \"Deprecate support for [ml2] tenant_network_types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eaf3669186fd36162bcb8f95883603e1e7041128"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-13 16:05:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-13 16:08:29.000000000","tz":540},"subject":"Add support for [DEFAULT] dnsmasq_txt_record","message":"Add support for [DEFAULT] dnsmasq_txt_record\n\nDepeneds-on: https://review.opendev.org/950486\nChange-Id: Idfe6a178f17e47643c5e7458df7404cbf4c14392\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d7923d17c0ff7d7d6de2939bf3072c03766f6097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d7923d17c0ff7d7d6de2939bf3072c03766f6097"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
