)]}'
{"id":"openstack%2Fpuppet-neutron~983370","triplet_id":"openstack%2Fpuppet-neutron~master~Ie3513d39cc0b64fa28b3b8f0288e7d4d89b1b3d8","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-06 15:40:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3513d39cc0b64fa28b3b8f0288e7d4d89b1b3d8","subject":"Add support for new options to control the OVN BFD settings","status":"MERGED","created":"2026-04-05 07:19:49.000000000","updated":"2026-04-06 15:41:39.000000000","submitted":"2026-04-06 15:40:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983370","meta_rev_id":"3953ee5a51caf74df05e44a97be73f4b9c71d6af","_number":983370,"virtual_id_number":983370,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-06 15:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2026-04-05 22:54:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-04-06 14:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2026-04-06 14:26:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-05 08:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-05 22:54:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-04-06 14:26:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"52bda5bfe63b1ab3a27227f265848b3f2e8835d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 07:19:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7d46e653b9e12a073077f5205d21fa188257e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 07:27:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"676ae1aa704f65d2feb2b5df50f2b256d944c687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 08:43:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e72dc0153cae4853add0a682169b82cf\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/266ceeb63005476ab8fcd6010943c22e : FAILURE in 2m 39s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f7adce77aa1428995d01b953eb2aaa7 : SUCCESS in 3m 01s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/21bcc7f1a10345c39af0cebcf45801c5 : SUCCESS in 7m 09s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81ccbc7a23fb456880297161c69cc807 : SUCCESS in 7m 47s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37c283ea0a3249c4bf2295fc6efd0b97 : SUCCESS in 11m 35s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a5d956400654c8a85c9002fd25208a5 : SUCCESS in 48m 39s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d56d116f2354b0e9a2c351357ba104e : SUCCESS in 52m 04s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40ac600ac9d648658a0d0ec3274f782c : FAILURE in 34m 27s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9d37005228c45b8afddcf2cb97c1701 : SUCCESS in 51m 06s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e6bbd806a764074b9e2bba4bb0ba64e : FAILURE in 43m 51s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c5e88d0747f49ee8b3d6dc32b5632c1 : SUCCESS in 52m 45s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f176e4345b5c465183e874f7bd927509 : SUCCESS in 1h 11m 06s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/075df2735cf547d6ba7622d6c320643c : SUCCESS in 44m 02s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6001344e9384126b0d6c1bc962a1942 : SUCCESS in 1h 04m 34s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/081943e211e7426b881df9c77c51d74a : SUCCESS in 1h 00m 28s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27eb2e3d8bf346bc8ca1da69101f3452 : SUCCESS in 1h 11m 45s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b41bf831a4d4473281f1846e7d5fdf9f : SUCCESS in 58m 27s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/141a9fcd79b7437aa8f09aa75056579b : SUCCESS in 32m 47s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f41f1191ac9a466cadd0bd0f854f1cbd : SUCCESS in 52m 09s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35b632b2f7654e90abedeaca54877d83 : SUCCESS in 1h 01m 09s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2903e99c38c048a6a746365fed9e5d46 : SUCCESS in 22m 54s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d0df96916404084bdef3e1e1a93bd29 : SUCCESS in 13m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2371a22d17a947c2b7665586a2dedde8 : SUCCESS in 10m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25263e0f9f9448eb84a9565f451a7cb2 : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":2},{"id":"64e24ab3344373c420423f8e724ec0b098fe1769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 11:39:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fd3cf497c84d07637622eeeb14b9da5bb02eb29a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 13:04:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4464ced95646a2ba4c20a0084467a2\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1bf8468d1ba4e98b0cbdea9ec5b176b : SUCCESS in 5m 24s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b4f6a1146df471f84a89d1c6ef95318 : SUCCESS in 3m 24s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/3abaad7765524d2489aac4c3bed61c7d : SUCCESS in 7m 38s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97ddd546926c4f2b8b83e06f4fde6778 : SUCCESS in 7m 25s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7695d3ca33324e658a871b1fd516fe2e : SUCCESS in 7m 22s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e57885e25ffe42899e8e5e21a8aaef1a : SUCCESS in 1h 00m 18s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50ff117c0a4e4701a86d83fc4e815bbd : SUCCESS in 1h 04m 19s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1abcc97e512c45aba45f70bc50a80593 : FAILURE in 34m 13s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8048aed346894cd9bc58a0e5c1503842 : SUCCESS in 50m 30s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f09f93d96ef54e4bba66c32552bf61de : FAILURE in 29m 40s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0abcf75bc9542adbe6c19548729a1cd : SUCCESS in 36m 12s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56a70f1b832c47ce9d356187c7700a0c : SUCCESS in 1h 20m 05s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2b244a647dc4e4bb45778b4cc7bea0e : SUCCESS in 1h 02m 10s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/00d193eb70564daca81912ab00baf3ba : SUCCESS in 36m 01s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca376d7ad35c42d1bde89a9bb78efcac : SUCCESS in 57m 25s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2892bb8292ed415b9b8a9be8f929d998 : SUCCESS in 51m 38s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c454649acd44a2a97cd2a9e06c6b689 : SUCCESS in 1h 03m 22s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/656147f3afb844c4a83142af052037d7 : SUCCESS in 47m 24s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82ae4b5a719b48cea91ef603508fe0a2 : SUCCESS in 1h 05m 02s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d76b65cd16a64c3c82cb49d1953d1d08 : SUCCESS in 33m 47s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/115a2877f9f04d0d884caf3c4ba34bbe : SUCCESS in 20m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a430b615d074b5caa47edd07106764a : SUCCESS in 17m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfd98afdd702412c936a1eceed6a0005 : SUCCESS in 10m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f9ddc48930c4b90a69d89c6b664a236 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"4d19eb1f9b9014c070c55ff605b59804863e45cc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-05 22:54:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bec193b5ea7f0882901c2759308a21c1dddcaff1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-04-06 14:26:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3db7d708eb8281b6aac58fa5ec4c018e17092952","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 14:26:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"33021968b2f26911d0417cd316455f601b59667e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 15:40:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d550d8ba6a3541cdb5693d34eb7f4bf0\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5ea1d0e577f477892ad9a0e4e0c57fc : SUCCESS in 3m 55s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c0e67ef623742038ee92695d1076912 : SUCCESS in 3m 09s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/23a6125a78c74fc4877ef5200d48cb7b : SUCCESS in 6m 28s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b38cc50ae8249e7980bcae2c2803264 : SUCCESS in 5m 26s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ae24cb5bc8741b3b96b9b4734a128b2 : SUCCESS in 1h 07m 32s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72743c238fed40caa5c763e37b29f5d0 : SUCCESS in 40m 36s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/870810adfd764d63ad32993450c2b66a : SUCCESS in 30m 00s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29362e2ae7e54bceab734fd28f8af164 : SUCCESS in 57m 51s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5076ce1737194f54812906ab81966d7f : SUCCESS in 1h 00m 15s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6e947bbb0a149098d34639f5f8f7ff3 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9257f31240044ee8e100f5d65a8ace8 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"54f2140cd5f10583757830c177a0255135cba8da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 15:40:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3953ee5a51caf74df05e44a97be73f4b9c71d6af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 15:41:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e4a532188be4a2ea326bbeb55027027\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2606e6712b7b490eb409526be74b8a22 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8222ff1144e85d2d7e3c6a75baeb6a5e5668c01d","revisions":{"a9496d571863ebeafa0a828982ac0b2da47cd320":{"kind":"REWORK","_number":1,"created":"2026-04-05 07:19:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/983370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/70/983370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/1"}}},"commit":{"parents":[{"commit":"de20e7ff452a7c4f44a24c74a9e2b0f315dad107","subject":"Merge \"Ubuntu/Debian: Adjust wsgi script path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de20e7ff452a7c4f44a24c74a9e2b0f315dad107"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:18:52.000000000","tz":540},"subject":"Add support for new options to control the OVN BFD settings","message":"Add support for new options to control the OVN BFD settings\n\nDepends-on: https://review.opendev.org/977209\nChange-Id: Ie3513d39cc0b64fa28b3b8f0288e7d4d89b1b3d8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a9496d571863ebeafa0a828982ac0b2da47cd320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a9496d571863ebeafa0a828982ac0b2da47cd320"}]},"branch":"refs/heads/master"},"6519f0c7fe487de005989e6017a138430c79e602":{"kind":"REWORK","_number":2,"created":"2026-04-05 07:27:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/983370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/70/983370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/2"}}},"commit":{"parents":[{"commit":"de20e7ff452a7c4f44a24c74a9e2b0f315dad107","subject":"Merge \"Ubuntu/Debian: Adjust wsgi script path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de20e7ff452a7c4f44a24c74a9e2b0f315dad107"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:27:22.000000000","tz":540},"subject":"Add support for new options to control the OVN BFD settings","message":"Add support for new options to control the OVN BFD settings\n\nDepends-on: https://review.opendev.org/977209\nChange-Id: Ie3513d39cc0b64fa28b3b8f0288e7d4d89b1b3d8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6519f0c7fe487de005989e6017a138430c79e602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6519f0c7fe487de005989e6017a138430c79e602"}]},"branch":"refs/heads/master"},"8222ff1144e85d2d7e3c6a75baeb6a5e5668c01d":{"kind":"REWORK","_number":3,"created":"2026-04-05 11:39:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/983370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/70/983370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/70/983370/3"}}},"commit":{"parents":[{"commit":"de20e7ff452a7c4f44a24c74a9e2b0f315dad107","subject":"Merge \"Ubuntu/Debian: Adjust wsgi script path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de20e7ff452a7c4f44a24c74a9e2b0f315dad107"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 11:39:21.000000000","tz":540},"subject":"Add support for new options to control the OVN BFD settings","message":"Add support for new options to control the OVN BFD settings\n\nDepends-on: https://review.opendev.org/977209\nChange-Id: Ie3513d39cc0b64fa28b3b8f0288e7d4d89b1b3d8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8222ff1144e85d2d7e3c6a75baeb6a5e5668c01d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8222ff1144e85d2d7e3c6a75baeb6a5e5668c01d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
