)]}'
{"id":"openstack%2Fpuppet-neutron~989852","triplet_id":"openstack%2Fpuppet-neutron~stable%2F2026.1~Id9ddf5a9dbd58819cd35999ca17a07563f8f9966","project":"openstack/puppet-neutron","branch":"stable/2026.1","hashtags":[],"change_id":"Id9ddf5a9dbd58819cd35999ca17a07563f8f9966","subject":"Fix typo in warning message","status":"MERGED","created":"2026-05-23 09:09:25.000000000","updated":"2026-05-23 17:51:24.000000000","submitted":"2026-05-23 17:51:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989852","cherry_pick_of_change":989800,"cherry_pick_of_patch_set":1,"meta_rev_id":"3537ee9750e147898dd46883f5ca894901e7eb91","_number":989852,"virtual_id_number":989852,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-23 17:51:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-05-23 16:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-05-23 16:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-23 10:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b0b5db8c7949491afb44a0856a272f2ec67ce52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-23 09:09:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"47b8873478a4ed041dfbde63a7e0a2f9776bc4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 10:33:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f86eb5c1058d49118ce2d5001a05ac0a\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2a7178258254fed9ce3180a3f787a40 : SUCCESS in 2m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06a55534f36e4893aeaa7d5f928d33c6 : SUCCESS in 2m 17s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/fdef08d4e4a949c1b5b4399af21dadce : SUCCESS in 7m 37s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fea10812a6e9420c9bb61957eeff2a18 : SUCCESS in 5m 28s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc044bfc18cf43c389c976f73e47ae5b : SUCCESS in 7m 04s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7045414e698440098e9feaf83aeb4e4 : FAILURE in 1h 11m 08s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbd650771f58412698e6c81e270db892 : FAILURE in 1h 16m 14s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d05afb99a9c24c739f09e66ff930e930 : FAILURE in 1h 06m 08s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd47bf454cbe474f8fddb06b18b9d6b0 : FAILURE in 59m 21s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78d43c7ecef24dab9925420e58c96377 : FAILURE in 1h 00m 18s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49c52a561d794cd3836526df565add2a : SUCCESS in 56m 35s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7208c02851d40e0a1a438902ea37b01 : SUCCESS in 1h 17m 49s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ce99aaf79824aceaeca891e14b07f11 : SUCCESS in 26m 44s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd977bfe725f4c5d88df8693dfcb7388 : SUCCESS in 31m 49s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/898ebb8f2f9a4be7ae18dd62c84167df : SUCCESS in 46m 08s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fda66ada6a09489f943bf686b403c002 : SUCCESS in 59m 49s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b003ca0df9342d6a2f47238be029035 : SUCCESS in 1h 00m 35s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e158c6ad4588465da80069547cf45bc6 : SUCCESS in 28m 21s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7b51a2fc1e14aa690ee5d5bffc5aca4 : SUCCESS in 58m 20s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/81a7e14f942a42f387a92cf4c23fab4b : SUCCESS in 31m 00s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/89faa5dc7126479eabd22ccd61689084 : SUCCESS in 10m 17s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c38bacc3e944bae8295d5d1d08a1579 : SUCCESS in 10m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48834dc7813b4946955483c922125141 : SUCCESS in 13m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/199bb34774604a4183e4236eb42fbe19 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"115c91dbd6dba4aae83ec7b61430209e8e63820a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-23 16:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fceb978a8eb60cf988a240b52d8efe4a2c0c2a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 16:18:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7e643d4ef2fd5056dd67ec7cefada92da07fadd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 17:51:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f77d8e9b3147cca3b8f6545ab7bb17\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e57b1bece0d246cea165ea0a374980fd : SUCCESS in 3m 43s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b497e27555b4bad86f6640f68984f00 : SUCCESS in 2m 42s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/9e6f2a14e5ef4ca8bfe0b62e0d516e16 : SUCCESS in 6m 50s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbf0e66e6f5748a684e682ab7579d885 : SUCCESS in 11m 33s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fbfe6e10bb14e3ca0a32b682b38f658 : SUCCESS in 1h 08m 02s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b23247501544f7b93b3e44af00ee87a : SUCCESS in 1h 25m 05s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40f65e51f0a34115bf15543468b27cf9 : SUCCESS in 59m 10s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c9dd7fdb786431eb061fa39b3f3db2a : SUCCESS in 55m 16s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/829ec6c0cbe64c2fb3387f10c6150e4f : SUCCESS in 55m 41s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6df6bc7bc02e413aab1114fe79c6a92a : SUCCESS in 19m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6630f8b2afaa46fca2ec33acabea543b : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3537ee9750e147898dd46883f5ca894901e7eb91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 17:51:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebfa6b3e511a209af78f323afcb9b81d8d7eb228","revisions":{"ebfa6b3e511a209af78f323afcb9b81d8d7eb228":{"kind":"REWORK","_number":1,"created":"2026-05-23 09:09:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/989852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/989852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/989852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/989852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/989852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/989852/1"}}},"commit":{"parents":[{"commit":"fe5bd47057ad924a7e5e0e3411e4b51bdb93800b","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fe5bd47057ad924a7e5e0e3411e4b51bdb93800b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-22 17:53:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-23 09:09:25.000000000","tz":0},"subject":"Fix typo in warning message","message":"Fix typo in warning message\n\nChange-Id: Id9ddf5a9dbd58819cd35999ca17a07563f8f9966\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit db8f9b75af754b87ae57fbe16910dc50550caae0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ebfa6b3e511a209af78f323afcb9b81d8d7eb228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ebfa6b3e511a209af78f323afcb9b81d8d7eb228"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"ours":"fe5bd47057ad924a7e5e0e3411e4b51bdb93800b","theirs":"db8f9b75af754b87ae57fbe16910dc50550caae0","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
