)]}'
{"id":"openstack%2Fpuppet-nova~1004752","triplet_id":"openstack%2Fpuppet-nova~master~I871dc1e2eca742f70354501be07469d5125a633d","project":"openstack/puppet-nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 16:32:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I871dc1e2eca742f70354501be07469d5125a633d","subject":"Deprecate support for [database] mysql_enable_nbd","status":"MERGED","created":"2026-09-09 02:53:30.000000000","updated":"2026-09-10 16:33:29.000000000","submitted":"2026-09-10 16:32:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004752","meta_rev_id":"d4c7b58bfbba6b1edee27ddba083502f2bfd83b7","_number":1004752,"virtual_id_number":1004752,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 16:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2026-09-10 10:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-09-10 14:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2026-09-10 14:50:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 04:54:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 10:54:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-09-10 14:50:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"1fced5c3199b02a9b9894e63813e1864477db74b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:53:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7e827e20c79815abf5784d6524b6c10dbac415d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 04:54:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f438d85fe9414a85acc8f87f16720bcb\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/696bccf888e541fb82b50ec8554922a5 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a23f72b7f73a478e939bba7d47211826 : SUCCESS in 3m 41s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e64a4e852a8d4c259bb5c56670ec92ef : FAILURE in 8m 26s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed4053a9f2d94b6899a25ce6ed24c760 : FAILURE in 5m 30s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c967dc6fba08403988df97df42be7380 : FAILURE in 6m 02s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af6851a792144dae8ff220afe4c16c3e : SUCCESS in 50m 50s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a08f768c7544ed7a6486ae87f9a2d9c : FAILURE in 1h 15m 56s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88b5bc789ad54ee4ac9c04aaf4136108 : FAILURE in 1h 17m 44s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3074d9ac2a754a199a484ef1dfdb12c2 : SUCCESS in 48m 53s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5c87ba3dd654f7c92cedb028d789eda : FAILURE in 51m 54s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbf237d35d4746d0bdcebf3aa7cb49f6 : SUCCESS in 31m 57s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78493be37a444b8dbe3ccfad810e8875 : SUCCESS in 1h 24m 01s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35dab2620b02475f9f5ef29cbc2836f6 : SUCCESS in 1h 07m 04s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf3f24628e0b47338b91b6e8404d5847 : SUCCESS in 59m 04s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be3445c72e9b4e7ca2190fca01763369 : SUCCESS in 1h 08m 13s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23f80ebd37064567ab46431094307b17 : SUCCESS in 1h 04m 01s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b096399747343729b5618f0db7a47c9 : SUCCESS in 1h 16m 47s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b82c07fbfc048168d7d190a83be7ac7 : SUCCESS in 33m 54s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11cc32f346614d4780e0791437998fc8 : SUCCESS in 1h 01m 43s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/451fdd4affe14af18fdb80ba78bfe340 : SUCCESS in 54m 25s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/37be7e6d0c574ce6981c9e4a6be1dc02 : FAILURE in 9m 34s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd2a422d870747358cdb51d21e9caef7 : SUCCESS in 24m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7187d71da45e4e8381a0a35b34a390c8 : SUCCESS in 24m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c68d863dd9e44c0d8c0597bd794b6776 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b30f43a27177487652777439e07adc99fe05b2d4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 16:19:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c87505228786c4a9a45d2202d9b5a0cfd7be760a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:01:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56025c6164dc4e7aad168e6824ac8458\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b212145f4e240acbab30a4b6da159a8 : SUCCESS in 2m 05s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fe7be52e71145229faa36b67afc4a19 : SUCCESS in 4m 02s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/312909af05cb4cad9ba646f922b6e871 : SUCCESS in 14m 40s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09958c6ab4944e35b4f651afccca7a88 : SUCCESS in 11m 27s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78324b7a1d374530ae331ae3bd2340eb : SUCCESS in 11m 32s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb23c9c9c0bc465aa8bd97ebb1deb77d : SUCCESS in 1h 11m 50s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea3959e20202442a93b9397cbad15fed : FAILURE in 1h 18m 18s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8565e5f6c63e4bec9e638d6f21cac842 : FAILURE in 1h 20m 02s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cee829a1729f4744bef5d382ec601b2a : SUCCESS in 49m 07s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b61be2410e24fa1bcdf50d2fbface0b : FAILURE in 1h 06m 38s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d6eb75175c84021aa17bc9fbb6f1436 : SUCCESS in 44m 10s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8b1bf40f0a543ba817fd7bcebb61af7 : SUCCESS in 1h 31m 07s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18d7c6f5d86d4f339dc37133b42585f0 : SUCCESS in 1h 00m 57s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66ebd581956247f7b15fd986eb9d1bcc : SUCCESS in 1h 01m 55s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/613c2104a80c4d5daad1a6f328b01125 : SUCCESS in 33m 16s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e049e5609e9e4d94b19861f2f569fed3 : SUCCESS in 1h 06m 31s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5857c39f01454561895a37603d7e776d : SUCCESS in 1h 20m 49s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d811667c0cb40f382d91954f839b2f6 : SUCCESS in 31m 21s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82a7602379094295898668990880bcd4 : SUCCESS in 55m 37s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22e5fbc868a943c7b6c7a36365f2358a : SUCCESS in 59m 10s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2852815961dc4261bb923924bd6f2d1f : SUCCESS in 28m 31s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23d0333374c64ecabf5fb19ef5c5d7d2 : SUCCESS in 26m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/687dbe5dd27d49b7b0e4679b84cd788d : SUCCESS in 12m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae7da06c9a8345d6a025d8a0dca663b8 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"56b4a9dc1519fe3b76b9c4d6865457a1663b8590","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-10 10:54:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b427d6d12dec7dcf777f2c463ecacafe92032e86","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-09-10 14:50:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ebb3d575efa67191db94d39dda73b62f719a93a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:57:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15e8279da5428538a42b0cb6f09e43498b15df12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:32:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0705555cdc51477488cfec9fe2d6e823\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/272c742a642d49aa9fdd372d8c4ffc51 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54627ca89238452698f99b0a9470776c : SUCCESS in 3m 39s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/7b05296b70a04b19b1c44679d28c2d67 : SUCCESS in 9m 18s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a9d62d7fbeb4e8182b934692e18fb2f : SUCCESS in 15m 49s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b74ae643504481881b1475f045eaf0c : SUCCESS in 1h 11m 42s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b4992dd1f2543c9b2bc170a55276aaf : SUCCESS in 1h 28m 35s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5326fbccba2341ab89ed81e2ab173a46 : SUCCESS in 1h 08m 34s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa5b34e4a81f4d1199b776b9004853d9 : SUCCESS in 1h 06m 25s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/45a20c14c8e04d8d9977428a2a4711cf : SUCCESS in 1h 05m 57s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5ef0c9a33f2452697925931ce65f1df : SUCCESS in 29m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8b918d66896443c902f2ba478c8734f : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"dcf03840614b2c54409ed53e882174343eef96b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:32:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4c7b58bfbba6b1edee27ddba083502f2bfd83b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:33:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79c170ac6f1e4b608376887f088e5e02\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53a3a065d71a40809f87395ef1b654a3 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27c7e8c6cde9bff9aa4d2f5000c07fac5c61bf2b","revisions":{"27c7e8c6cde9bff9aa4d2f5000c07fac5c61bf2b":{"kind":"REWORK","_number":1,"created":"2026-09-09 02:53:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1004752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/52/1004752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/52/1004752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/52/1004752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/52/1004752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/52/1004752/1"}}},"commit":{"parents":[{"commit":"8b065da6e622c08dab04ab9f85b4eb5ee247dec5","subject":"Merge \"Remove PostgreSQL support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8b065da6e622c08dab04ab9f85b4eb5ee247dec5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:53:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:53:20.000000000","tz":540},"subject":"Deprecate support for [database] mysql_enable_nbd","message":"Deprecate support for [database] mysql_enable_nbd\n\nDepends-on: https://review.opendev.org/1004735\nChange-Id: I871dc1e2eca742f70354501be07469d5125a633d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/27c7e8c6cde9bff9aa4d2f5000c07fac5c61bf2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/27c7e8c6cde9bff9aa4d2f5000c07fac5c61bf2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
