)]}'
{"id":"openstack%2Fpuppet-nova~131710","triplet_id":"openstack%2Fpuppet-nova~master~Ib060bff4b1f547b4db255969bc8cb3119a85ba6d","project":"openstack/puppet-nova","branch":"master","topic":"bug/1378962","hashtags":[],"change_id":"Ib060bff4b1f547b4db255969bc8cb3119a85ba6d","subject":"Add host_uuid to libvirtd.conf","status":"MERGED","created":"2014-10-29 11:30:06.000000000","updated":"2016-01-26 15:22:19.000000000","submitted":"2016-01-26 15:22:19.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7cc5ff38364d57888c82e7964b6617754dcc5a44","_number":131710,"virtual_id_number":131710,"owner":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":2,"date":"2016-01-26 15:22:19.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"value":0,"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},{"value":0,"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"value":0,"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"all":[{"value":0,"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},{"value":2,"date":"2016-01-25 12:38:08.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2016-01-26 15:03:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2016-01-25 12:55:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"value":0,"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},{"value":0,"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"value":0,"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":1,"date":"2016-01-26 15:03:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"value":0,"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},{"value":0,"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"value":0,"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-12-15 10:47:09.000000000","updated_by":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"reviewer":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"state":"REVIEWER"},{"updated":"2014-12-29 13:43:17.000000000","updated_by":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"reviewer":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"state":"REVIEWER"},{"updated":"2015-12-01 06:05:41.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2015-12-08 22:47:35.000000000","updated_by":{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},"reviewer":{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},"state":"REVIEWER"},{"updated":"2015-12-21 13:49:44.000000000","updated_by":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"reviewer":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"state":"REVIEWER"},{"updated":"2015-12-22 15:03:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2015-12-22 15:09:05.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2016-01-07 04:18:32.000000000","updated_by":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"reviewer":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"state":"REVIEWER"},{"updated":"2016-01-25 12:38:08.000000000","updated_by":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"reviewer":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"state":"REVIEWER"},{"updated":"2016-01-25 12:55:35.000000000","updated_by":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"reviewer":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"state":"REVIEWER"},{"updated":"2016-01-26 15:03:40.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-01-26 15:22:19.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1701b5b609041364973c511cc1533f241a542415","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-10-29 11:30:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de8aa278ff0cf0bebed7700cb36e6b9019e1aaf6","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-10-29 11:31:07.000000000","message":"Patch Set 2: Commit message was updated","accounts_in_message":[],"_revision_number":2},{"id":"93f2ab37963090b3ce948b913c292307f154d0d7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 11:40:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/2/check/gate-puppet-nova-puppet-lint/92fbbd6 : FAILURE in 31s\n- gate-puppet-nova-puppet-syntax http://logs.openstack.org/10/131710/2/check/gate-puppet-nova-puppet-syntax/65d7acc : SUCCESS in 27s\n- gate-puppet-nova-puppet-unit-3.1 http://logs.openstack.org/10/131710/2/check/gate-puppet-nova-puppet-unit-3.1/95ad1c2 : SUCCESS in 2m 36s\n- gate-puppet-nova-puppet-unit-3.2 http://logs.openstack.org/10/131710/2/check/gate-puppet-nova-puppet-unit-3.2/7a6a279 : SUCCESS in 2m 11s\n- gate-puppet-nova-puppet-unit-3.3 http://logs.openstack.org/10/131710/2/check/gate-puppet-nova-puppet-unit-3.3/8685355 : SUCCESS in 2m 51s\n- gate-puppet-nova-puppet-unit-3.4 http://logs.openstack.org/10/131710/2/check/gate-puppet-nova-puppet-unit-3.4/25ecc01 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b06bc6ba109af441371f2d89343763c9b875880b","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-10-29 11:58:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"70ec36fe93b80f3ff5c22892329975126717893b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 12:06:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/3/check/gate-puppet-nova-puppet-lint/61c750a : SUCCESS in 36s\n- gate-puppet-nova-puppet-syntax http://logs.openstack.org/10/131710/3/check/gate-puppet-nova-puppet-syntax/20c8e55 : SUCCESS in 9s\n- gate-puppet-nova-puppet-unit-3.1 http://logs.openstack.org/10/131710/3/check/gate-puppet-nova-puppet-unit-3.1/e0ed058 : SUCCESS in 2m 06s\n- gate-puppet-nova-puppet-unit-3.2 http://logs.openstack.org/10/131710/3/check/gate-puppet-nova-puppet-unit-3.2/0a9d80b : SUCCESS in 3m 34s\n- gate-puppet-nova-puppet-unit-3.3 http://logs.openstack.org/10/131710/3/check/gate-puppet-nova-puppet-unit-3.3/14cd730 : SUCCESS in 2m 51s\n- gate-puppet-nova-puppet-unit-3.4 http://logs.openstack.org/10/131710/3/check/gate-puppet-nova-puppet-unit-3.4/af8d367 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"076d1ae9ea1643e18f793fd8698039e7fd7ecf2d","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-11-07 15:50:29.000000000","message":"Patch Set 4: Commit message was updated","accounts_in_message":[],"_revision_number":4},{"id":"cd97b3ec5b7697f6824f85f6f8667a5ffb2f8e47","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-07 15:58:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/4/check/gate-puppet-nova-puppet-lint/b4f7a8a : SUCCESS in 34s\n- gate-puppet-nova-puppet-syntax http://logs.openstack.org/10/131710/4/check/gate-puppet-nova-puppet-syntax/30b4f03 : SUCCESS in 22s\n- gate-puppet-nova-puppet-unit-3.1 http://logs.openstack.org/10/131710/4/check/gate-puppet-nova-puppet-unit-3.1/dbf9fed : SUCCESS in 2m 38s\n- gate-puppet-nova-puppet-unit-3.2 http://logs.openstack.org/10/131710/4/check/gate-puppet-nova-puppet-unit-3.2/bc0de7c : SUCCESS in 2m 43s\n- gate-puppet-nova-puppet-unit-3.3 http://logs.openstack.org/10/131710/4/check/gate-puppet-nova-puppet-unit-3.3/58dc8cc : SUCCESS in 2m 17s\n- gate-puppet-nova-puppet-unit-3.4 http://logs.openstack.org/10/131710/4/check/gate-puppet-nova-puppet-unit-3.4/5e94d98 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"d954317e230e1cf39a106b8efee2d1b7df33a23c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2014-11-12 22:27:12.000000000","message":"Patch Set 4: Code-Review-1\n\nThanks for this addition, please add unit tests.","accounts_in_message":[],"_revision_number":4},{"id":"5ed0e3621c73f8e6c55059e650fde98eab6a6813","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-12-02 16:23:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0544836db0ad13c1bf26d95211dc16bda52f87d0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-02 16:34:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/5/check/gate-puppet-nova-puppet-lint/70f8054 : SUCCESS in 35s\n- gate-puppet-nova-puppet-syntax http://logs.openstack.org/10/131710/5/check/gate-puppet-nova-puppet-syntax/2fc50b1 : SUCCESS in 9s\n- gate-puppet-nova-puppet-unit-3.3 http://logs.openstack.org/10/131710/5/check/gate-puppet-nova-puppet-unit-3.3/b79cc25 : FAILURE in 2m 40s\n- gate-puppet-nova-puppet-unit-3.4 http://logs.openstack.org/10/131710/5/check/gate-puppet-nova-puppet-unit-3.4/97e1cf4 : FAILURE in 2m 37s\n- gate-puppet-nova-puppet-unit-3.6 http://logs.openstack.org/10/131710/5/check/gate-puppet-nova-puppet-unit-3.6/974f5d3 : FAILURE in 1m 58s\n- gate-puppet-nova-puppet-unit-3.7 http://logs.openstack.org/10/131710/5/check/gate-puppet-nova-puppet-unit-3.7/f5a86a1 : FAILURE in 2m 32s","accounts_in_message":[],"_revision_number":5},{"id":"76691962e55e5086e2b403e3169094621788e9a4","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-12-02 16:41:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8413a6f63b8d7ec9353177cc36063827de4b3049","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-02 16:44:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/6/check/gate-puppet-nova-puppet-lint/4630137 : SUCCESS in 35s\n- gate-puppet-nova-puppet-syntax http://logs.openstack.org/10/131710/6/check/gate-puppet-nova-puppet-syntax/b68cc36 : SUCCESS in 10s\n- gate-puppet-nova-puppet-unit-3.3 http://logs.openstack.org/10/131710/6/check/gate-puppet-nova-puppet-unit-3.3/1535b3a : SUCCESS in 2m 00s\n- gate-puppet-nova-puppet-unit-3.4 http://logs.openstack.org/10/131710/6/check/gate-puppet-nova-puppet-unit-3.4/3b64a37 : SUCCESS in 2m 39s\n- gate-puppet-nova-puppet-unit-3.6 http://logs.openstack.org/10/131710/6/check/gate-puppet-nova-puppet-unit-3.6/f06b537 : SUCCESS in 2m 03s\n- gate-puppet-nova-puppet-unit-3.7 http://logs.openstack.org/10/131710/6/check/gate-puppet-nova-puppet-unit-3.7/9fec517 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":6},{"id":"729b8ef48af7610877dba53dd1f026283bf232c6","author":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"date":"2014-12-15 10:47:09.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ab388b481cd641aee15dfb8c4d1b664f7569e853","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-12-15 10:51:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9a9ce7b7a603454a996d16a50aa55e5e75de7566","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2014-12-23 03:50:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9b56136a44fde2fc78807ccdaad4ef0a464beb27","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-12-23 13:29:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"137ee23598c43d51ce9ab2d6b0d4513133ddaece","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2014-12-23 18:16:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3630d4996b15e9451459603b3d2b3e7e59eb2a4c","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2014-12-23 19:12:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d0510b03b33a749c137417b7f16fafb4760944c0","author":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"date":"2014-12-29 13:43:17.000000000","message":"Patch Set 6: Code-Review-1\n\nLibvirt already provides default mechanisms to obtain a valid host_uuid. To not break existing setups, you should add a parameter to allow creating a different host_uuid than the default provided by libvirt and do nothing as a default.","accounts_in_message":[],"_revision_number":6},{"id":"1ebdb9dc7ab1e98f018b2d8216d3e23ffb15b0b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-21 03:04:12.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"8b1aa0d857ce68880b9468a0c93f1ff87220faed","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2015-11-26 10:58:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2cb45adef17cdacf8abbdf75697dc71cadc2e6fb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-11-30 10:38:22.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"db50e2c94f9189e87942633787d8f0bb9c519b41","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-11-30 10:45:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4633441a36da3d1da7bb8da3a545f4bcfe91d70a","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2015-11-30 11:12:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fca68520bc9f985bf3bd94c1d153804af435e463","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-30 11:48:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-lint/e242241/ : SUCCESS in 47s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/092f103/ : SUCCESS in 2m 50s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/9e62d52/ : SUCCESS in 2m 48s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/279ace6/ : SUCCESS in 7m 56s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/1bf2189/ : SUCCESS in 6m 21s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/8d0cdd8/ : SUCCESS in 12m 14s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/20a18f4/ : SUCCESS in 10m 39s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/4fd7179/ : SUCCESS in 11m 37s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/6b4df59/ : SUCCESS in 8m 00s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/881383a/ : SUCCESS in 9m 12s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/d0cf8a0/ : SUCCESS in 10m 02s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/6d3b0b8/ : SUCCESS in 11m 18s\n- gate-puppet-openstack-integration-scenario001-dsvm-trusty http://logs.openstack.org/10/131710/7/check/gate-puppet-openstack-integration-scenario001-dsvm-trusty/49597c1/ : SUCCESS in 25m 43s (non-voting)\n- gate-puppet-openstack-integration-scenario001-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-openstack-integration-scenario001-dsvm-centos7/02fa93a/ : SUCCESS in 19m 42s (non-voting)\n- gate-puppet-openstack-integration-scenario002-dsvm-trusty http://logs.openstack.org/10/131710/7/check/gate-puppet-openstack-integration-scenario002-dsvm-trusty/47ebf45/ : SUCCESS in 25m 23s (non-voting)\n- gate-puppet-openstack-integration-scenario002-dsvm-centos7 http://logs.openstack.org/10/131710/7/check/gate-puppet-openstack-integration-scenario002-dsvm-centos7/8af277f/ : SUCCESS in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"82ac74f2ceafa37e0f792b8deabc237e2fce284d","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-11-30 11:57:59.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b7cf355f717832b66609b06e040ec281f581c21e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-01 04:02:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f21-nonha http://logs.openstack.org/10/131710/7/check-tripleo/gate-tripleo-ci-f21-nonha/b187108/ : SUCCESS in 1h 39m 05s","accounts_in_message":[],"_revision_number":7},{"id":"d537641c2fb19100d6a3718619a4843b897791cc","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2015-12-01 06:05:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"731f7e4de7109aa89aee8da5e6b0907b0d137d6e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-12-01 14:44:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fc15d1bdacc62e483f4dc6aed6eea2cad66b5cac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-08 19:39:06.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"fe6957d483aca525c8a4bdaf3255fc850a9962e6","author":{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},"date":"2015-12-08 22:47:35.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a2d6bfe5e4e20fbc5c0ac861557c2619ec964c82","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-12-09 09:42:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0bc9de174fefde40cd9c5451afe3f3c2c45835f2","author":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"date":"2015-12-21 13:49:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f568338b1903e14a752c35e630f2edab8711ddd4","author":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"date":"2015-12-21 13:57:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"66dea85abf0e5497a98cdf54a80e3d39c934ddc5","author":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"date":"2015-12-21 14:02:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c474d24f778ce23cb2c8a2859563fc7e9321a81c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-12-21 14:10:31.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"413d17f79b357c47bdb9788eeefc5a8718557604","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2015-12-21 14:21:48.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1bf3cdc3735ae321f058170589950c483a3b804a","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2015-12-21 14:22:34.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bff0ffbd88308e50af76047aa8d07a2a2e3501ea","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2015-12-21 14:35:02.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d600ad5822bc1a50cbabe089f394a124e0dc08b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-21 14:54:25.000000000","message":"Patch Set 13:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/10/131710/13/check-tripleo/gate-tripleo-ci-f22-nonha/5455cd4/ : FAILURE in 19m 14s","accounts_in_message":[],"_revision_number":13},{"id":"f44aa879c872193975239930b3f83bbd690a281c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-21 15:11:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-lint/67a3718/ : FAILURE in 45s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/09f1059/ : SUCCESS in 2m 57s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/4068eec/ : SUCCESS in 3m 07s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/bec603e/ : FAILURE in 8m 50s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/45a9a3b/ : FAILURE in 8m 34s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/b8535e3/ : FAILURE in 11m 48s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/15dde18/ : FAILURE in 10m 59s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/4a71218/ : FAILURE in 11m 26s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/4a14d5b/ : FAILURE in 9m 10s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/725f0eb/ : FAILURE in 10m 35s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/c876746/ : SUCCESS in 9m 45s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/6def9e2/ : FAILURE in 4m 38s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/10/131710/13/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/1c8fbb4/ : SUCCESS in 22m 39s (non-voting)\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/33f8aba/ : FAILURE in 1m 31s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/10/131710/13/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/7db5e9b/ : SUCCESS in 35m 45s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/13/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/625eaf6/ : FAILURE in 1m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fe8be894c197d84fb38ec641687eca2df125010a","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2015-12-21 21:52:39.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a645d675fbaced908f202ca1b842b4e1bafbb406","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-21 22:14:18.000000000","message":"Patch Set 14:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/10/131710/14/check-tripleo/gate-tripleo-ci-f22-nonha/571e7cd/ : FAILURE in 21m 32s","accounts_in_message":[],"_revision_number":14},{"id":"c767f16f9f0c9e6eb47512f0b24b9fe683f269ff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-21 22:20:50.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-lint/3531f60/ : SUCCESS in 50s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/d6be4e8/ : SUCCESS in 3m 03s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/223f809/ : SUCCESS in 3m 04s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/e3295cf/ : FAILURE in 7m 51s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/92d9a15/ : FAILURE in 8m 47s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/29c7f74/ : SUCCESS in 10m 51s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/460d83e/ : SUCCESS in 11m 57s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/fe0e668/ : SUCCESS in 12m 38s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/eeacb49/ : SUCCESS in 8m 40s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/a02fb1f/ : SUCCESS in 10m 16s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/f55dcd1/ : SUCCESS in 13m 39s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/3ac7676/ : FAILURE in 4m 35s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/10/131710/14/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/758a4e5/ : SUCCESS in 25m 18s (non-voting)\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/e0a11f4/ : FAILURE in 1m 37s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/10/131710/14/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/d012555/ : FAILURE in 27m 20s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/14/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/097b5b3/ : FAILURE in 1m 33s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9cfa7ee7a119ac2b30d08c797a8d59427235b523","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2015-12-21 23:36:19.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"6ec157bdcd5421ecb830fed0f38484c16c74415f","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2015-12-22 09:24:31.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d599a3bd1c6e42c729033b4fc35faf7b4192dacf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-22 10:21:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-lint/882b52e/ : SUCCESS in 48s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/b49f30f/ : SUCCESS in 2m 50s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/94570a1/ : SUCCESS in 2m 47s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/390b2c0/ : SUCCESS in 7m 56s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/10e6704/ : SUCCESS in 8m 06s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/9cd4835/ : SUCCESS in 12m 33s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/47728ad/ : SUCCESS in 12m 33s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/40e3cdd/ : SUCCESS in 11m 58s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/5aabe2a/ : SUCCESS in 9m 32s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/4f319c4/ : SUCCESS in 11m 31s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/0e808a7/ : SUCCESS in 11m 42s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/c3f4b90/ : SUCCESS in 10m 14s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/10/131710/15/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/5a94949/ : SUCCESS in 30m 25s (non-voting)\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/4a62345/ : SUCCESS in 28m 05s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/10/131710/15/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/81e53f2/ : SUCCESS in 30m 36s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/15/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/4b5e975/ : SUCCESS in 26m 00s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"390875e53f18cd3da270341344199d5e4605a4b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-22 10:47:06.000000000","message":"Patch Set 15:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/10/131710/15/check-tripleo/gate-tripleo-ci-f22-nonha/647e5cd/ : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":15},{"id":"f09ad5f849706127236cf24437891d1f043e5ea8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2015-12-22 15:03:52.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"67663108b9c0484584abb914dffeef15bfd555b5","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-12-22 15:09:05.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"4681e98b7ab948e56a4febd90eacb61f0b2ce633","author":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"date":"2015-12-23 11:45:59.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"98b4689aedebe9cbc3138d1c7cd3c390d49d689f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-12-30 09:03:06.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"e892d45afd8ad2e6df760c290a8aa545c510b378","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2016-01-07 04:18:32.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f92dd64821c9c5297fd044fe2945c7343fec123c","author":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"date":"2016-01-08 11:56:43.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a5b471165c73b339f39646f790cf53fcf9344e0a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-08 12:35:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-lint/3f24f58/ : SUCCESS in 54s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/cc0f4fc/ : SUCCESS in 3m 03s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/05f9823/ : SUCCESS in 3m 04s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/61ec098/ : SUCCESS in 8m 05s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/1c31d8c/ : SUCCESS in 8m 40s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/c2ac710/ : SUCCESS in 10m 49s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/d899753/ : SUCCESS in 11m 37s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/2c4b917/ : SUCCESS in 11m 14s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/4b167d9/ : SUCCESS in 9m 15s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/f87bd91/ : SUCCESS in 10m 54s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/2b55c03/ : SUCCESS in 8m 54s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/11de7cf/ : SUCCESS in 11m 36s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/660bc18/ : FAILURE in 23m 48s (non-voting)\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/12220b8/ : FAILURE in 13m 01s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/27788a8/ : SUCCESS in 29m 07s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/a73b218/ : SUCCESS in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0ed0e5c3cf831f0bb4d2bfddbaf7aed1dc7326df","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-08 13:31:45.000000000","message":"Patch Set 16:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/10/131710/16/check-tripleo/gate-tripleo-ci-f22-nonha/3153081/ : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":16},{"id":"9f17f8674e9b22de34e5455a70d422f47b3092cc","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-01-25 11:38:18.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"626b392d4ca3033f78de43a521449d2ffa4fc0b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-25 12:21:07.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-lint/760c17a/ : SUCCESS in 52s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/0ad4337/ : SUCCESS in 3m 00s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/7aa9daa/ : SUCCESS in 2m 55s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/d4d3a5b/ : SUCCESS in 7m 33s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/e8184f3/ : SUCCESS in 7m 41s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/2501ce5/ : SUCCESS in 11m 06s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/f8ed7f2/ : SUCCESS in 11m 24s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/3ffc66c/ : SUCCESS in 11m 24s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/0bde0bc/ : SUCCESS in 9m 01s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/487096b/ : SUCCESS in 10m 11s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/b555041/ : SUCCESS in 8m 34s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/c892948/ : SUCCESS in 11m 53s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/a060729/ : SUCCESS in 39m 43s (non-voting)\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/7bb3a49/ : SUCCESS in 31m 54s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/a300211/ : SUCCESS in 30m 05s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/10/131710/16/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/86697f9/ : SUCCESS in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6082ac14e4711ab2531fcd718f730b14954f3438","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-01-25 12:38:08.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"1494d1d8feee9623c123bebeb4488158bd51942b","author":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"date":"2016-01-25 12:55:35.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"8665cff1a81d7e0498b895ae2b93ce824d8c3017","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-25 12:57:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/10/131710/16/check-tripleo/gate-tripleo-ci-f22-nonha/9d9370f/ : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":16},{"id":"e59517a08a9ac986314e58120da1751894837157","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-01-26 15:03:40.000000000","message":"Patch Set 16: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"2fc9964ae671171cadae5f0abca864ee968ce1b1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-26 15:05:07.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":16},{"id":"691f50152859ee4a72dcb415b5247bb36e03ef80","date":"2016-01-26 15:22:19.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":16},{"id":"192e3ed31be15342821d860add15e1d5278a8f26","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-26 15:22:19.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-lint/78ff864/ : SUCCESS in 55s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/4b2109a/ : SUCCESS in 3m 20s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/88dbcf9/ : SUCCESS in 3m 07s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/973be68/ : SUCCESS in 7m 57s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/09901a4/ : SUCCESS in 8m 21s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/986c20f/ : SUCCESS in 13m 21s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/6b1326f/ : SUCCESS in 12m 02s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/00accb4/ : SUCCESS in 11m 58s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/76b6052/ : SUCCESS in 9m 16s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/0ad07d0/ : SUCCESS in 14m 13s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/10/131710/16/gate/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/6d9792f/ : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"8bf30c41e90f1571630e8b77f3c97f85328f9e5f","revisions":{"a23b481ffdb737461cc5eb4743c5cf636259ac33":{"kind":"REWORK","_number":1,"created":"2014-10-29 11:30:06.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/1"}}},"commit":{"parents":[{"commit":"648c9e85830204995c24ae43d15efe278b9fa56b","subject":"Switch to metadata.json for Puppet forge releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/648c9e85830204995c24ae43d15efe278b9fa56b"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug:\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a23b481ffdb737461cc5eb4743c5cf636259ac33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a23b481ffdb737461cc5eb4743c5cf636259ac33"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"648c9e85830204995c24ae43d15efe278b9fa56b","is_merged_in_target_branch":true,"change_id":"Iac5654750e46860888c0a035ed22f75f5f7bed3a","change_number":124581,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f30840e0f575bb4bf984b22a869d8134cd9334f3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2014-10-29 11:31:07.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/2"}}},"commit":{"parents":[{"commit":"648c9e85830204995c24ae43d15efe278b9fa56b","subject":"Switch to metadata.json for Puppet forge releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/648c9e85830204995c24ae43d15efe278b9fa56b"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:31:07.000000000","tz":0},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f30840e0f575bb4bf984b22a869d8134cd9334f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f30840e0f575bb4bf984b22a869d8134cd9334f3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"648c9e85830204995c24ae43d15efe278b9fa56b","is_merged_in_target_branch":true,"change_id":"Iac5654750e46860888c0a035ed22f75f5f7bed3a","change_number":124581,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"20f4261cdfc47dc8c9df89f7e427ee547eb2468e":{"kind":"REWORK","_number":3,"created":"2014-10-29 11:58:14.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/3"}}},"commit":{"parents":[{"commit":"648c9e85830204995c24ae43d15efe278b9fa56b","subject":"Switch to metadata.json for Puppet forge releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/648c9e85830204995c24ae43d15efe278b9fa56b"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:57:57.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug:\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/20f4261cdfc47dc8c9df89f7e427ee547eb2468e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/20f4261cdfc47dc8c9df89f7e427ee547eb2468e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"648c9e85830204995c24ae43d15efe278b9fa56b","is_merged_in_target_branch":true,"change_id":"Iac5654750e46860888c0a035ed22f75f5f7bed3a","change_number":124581,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f0818033aef74d51a8edaf89e7f3b66f865e08ae":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2014-11-07 15:50:29.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/4"}}},"commit":{"parents":[{"commit":"648c9e85830204995c24ae43d15efe278b9fa56b","subject":"Switch to metadata.json for Puppet forge releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/648c9e85830204995c24ae43d15efe278b9fa56b"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-11-07 15:50:29.000000000","tz":0},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f0818033aef74d51a8edaf89e7f3b66f865e08ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f0818033aef74d51a8edaf89e7f3b66f865e08ae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"648c9e85830204995c24ae43d15efe278b9fa56b","is_merged_in_target_branch":true,"change_id":"Iac5654750e46860888c0a035ed22f75f5f7bed3a","change_number":124581,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ecdbe2a7a5e80f4b4a303cad5b574f05eafc823c":{"kind":"REWORK","_number":5,"created":"2014-12-02 16:23:30.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/5"}}},"commit":{"parents":[{"commit":"30b6a9c4c7d9b662eb8e70b58a1ad63d1c4e89aa","subject":"Configure database parameters on the right nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/30b6a9c4c7d9b662eb8e70b58a1ad63d1c4e89aa"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-12-02 16:23:28.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/ecdbe2a7a5e80f4b4a303cad5b574f05eafc823c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/ecdbe2a7a5e80f4b4a303cad5b574f05eafc823c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"30b6a9c4c7d9b662eb8e70b58a1ad63d1c4e89aa","is_merged_in_target_branch":true,"change_id":"I42b9d2b1efb5856fed6550c25ac3142952690df1","change_number":135840,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ff057aefbab2a6393fee30997d290e486fa15d65":{"kind":"REWORK","_number":6,"created":"2014-12-02 16:41:02.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/6"}}},"commit":{"parents":[{"commit":"30b6a9c4c7d9b662eb8e70b58a1ad63d1c4e89aa","subject":"Configure database parameters on the right nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/30b6a9c4c7d9b662eb8e70b58a1ad63d1c4e89aa"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-12-02 16:40:52.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/ff057aefbab2a6393fee30997d290e486fa15d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/ff057aefbab2a6393fee30997d290e486fa15d65"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"30b6a9c4c7d9b662eb8e70b58a1ad63d1c4e89aa","is_merged_in_target_branch":true,"change_id":"I42b9d2b1efb5856fed6550c25ac3142952690df1","change_number":135840,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"40cde61fc22040a8882d63782d9493fa211a5954":{"kind":"REWORK","_number":7,"created":"2015-11-30 11:12:46.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/7"}}},"commit":{"parents":[{"commit":"8bb871e42101d13df933a3760645134fe5674d8c","subject":"Checkout openstack-integration from master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8bb871e42101d13df933a3760645134fe5674d8c"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2015-11-30 11:12:44.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/40cde61fc22040a8882d63782d9493fa211a5954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/40cde61fc22040a8882d63782d9493fa211a5954"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8bb871e42101d13df933a3760645134fe5674d8c","is_merged_in_target_branch":true,"change_id":"Id2da89a8cecb53fb8efbf9bf9f3ac72feef87595","change_number":250296,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2bdf2c50977d6ebbdacd5fbac0ac2dda8c1233d1":{"kind":"REWORK","_number":8,"created":"2015-12-21 13:49:44.000000000","uploader":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"ref":"refs/changes/10/131710/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/8"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Dmitry Ilyin","email":"dilyin@mirantis.com","date":"2015-12-21 13:49:13.000000000","tz":180},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/2bdf2c50977d6ebbdacd5fbac0ac2dda8c1233d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/2bdf2c50977d6ebbdacd5fbac0ac2dda8c1233d1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4f2a4856273cb0d7c6e2f6b3a302bce7b3d1d122":{"kind":"REWORK","_number":9,"created":"2015-12-21 13:57:54.000000000","uploader":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"ref":"refs/changes/10/131710/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/9"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Dmitry Ilyin","email":"dilyin@mirantis.com","date":"2015-12-21 13:57:35.000000000","tz":180},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/4f2a4856273cb0d7c6e2f6b3a302bce7b3d1d122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/4f2a4856273cb0d7c6e2f6b3a302bce7b3d1d122"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3f6b3517c998829039a3e5f26e57734c371a65ff":{"kind":"REWORK","_number":10,"created":"2015-12-21 14:02:10.000000000","uploader":{"_account_id":9037,"name":"Dmitry Ilyin","email":"idv1985@gmail.com","username":"dilyin"},"ref":"refs/changes/10/131710/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/10"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Dmitry Ilyin","email":"dilyin@mirantis.com","date":"2015-12-21 14:02:07.000000000","tz":180},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/3f6b3517c998829039a3e5f26e57734c371a65ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/3f6b3517c998829039a3e5f26e57734c371a65ff"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9b0ce652fa89603e2628d3f58e5e8bb73801bf61":{"kind":"REWORK","_number":11,"created":"2015-12-21 14:21:48.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/11"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2015-12-21 14:21:37.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/9b0ce652fa89603e2628d3f58e5e8bb73801bf61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/9b0ce652fa89603e2628d3f58e5e8bb73801bf61"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1a85a8c3fe92657121513f67306033ffe503dd5d":{"kind":"REWORK","_number":12,"created":"2015-12-21 14:22:34.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/12"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2015-12-21 14:22:26.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/1a85a8c3fe92657121513f67306033ffe503dd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/1a85a8c3fe92657121513f67306033ffe503dd5d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0fb3fb789a389209f733035a99568fec3ddbdcc0":{"kind":"REWORK","_number":13,"created":"2015-12-21 14:35:02.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/13"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2015-12-21 14:34:54.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/0fb3fb789a389209f733035a99568fec3ddbdcc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/0fb3fb789a389209f733035a99568fec3ddbdcc0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"743c0e920627344c29ac7c7877efaf2bd9ae885c":{"kind":"REWORK","_number":14,"created":"2015-12-21 21:52:39.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/14"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2015-12-21 21:52:20.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/743c0e920627344c29ac7c7877efaf2bd9ae885c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/743c0e920627344c29ac7c7877efaf2bd9ae885c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"61819b43869aba4af83e0fba3517033cc40972e3":{"kind":"REWORK","_number":15,"created":"2015-12-22 09:24:31.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/15"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2015-12-22 09:24:20.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/61819b43869aba4af83e0fba3517033cc40972e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/61819b43869aba4af83e0fba3517033cc40972e3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8bf30c41e90f1571630e8b77f3c97f85328f9e5f":{"kind":"REWORK","_number":16,"created":"2016-01-08 11:56:43.000000000","uploader":{"_account_id":13344,"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","username":"cvieri.lviv"},"ref":"refs/changes/10/131710/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/10/131710/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/10/131710/16"}}},"commit":{"parents":[{"commit":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","subject":"Merge \"Use auth_uri to get nova credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d93b9709af1786ad3b2401c2de2fef9f96bd6827"}]}],"author":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2014-10-29 11:20:18.000000000","tz":120},"committer":{"name":"Oleksiy Molchanov","email":"omolchanov@mirantis.com","date":"2016-01-08 11:56:09.000000000","tz":120},"subject":"Add host_uuid to libvirtd.conf","message":"Add host_uuid to libvirtd.conf\n\nAdd host_uuid generated by uuidgen to libvirtd.conf\n\nChange-Id: Ib060bff4b1f547b4db255969bc8cb3119a85ba6d\nCloses-Bug: 1378962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8bf30c41e90f1571630e8b77f3c97f85328f9e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8bf30c41e90f1571630e8b77f3c97f85328f9e5f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d93b9709af1786ad3b2401c2de2fef9f96bd6827","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
