)]}'
{"id":"openstack%2Fpuppet-nova~263957","triplet_id":"openstack%2Fpuppet-nova~stable%2Fliberty~Ibc5b9717d681d9a090777d4fe64b906c3c07ec5d","project":"openstack/puppet-nova","branch":"stable/liberty","topic":"puppet/liberty","hashtags":[],"change_id":"Ibc5b9717d681d9a090777d4fe64b906c3c07ec5d","subject":"Support of PyMySQL driver for MySQL backend","status":"MERGED","created":"2016-01-05 23:53:57.000000000","updated":"2016-01-06 11:03:16.000000000","submitted":"2016-01-06 11:03:16.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6c31a2a58849ca262a926b0b197c401e8deb952","_number":263957,"virtual_id_number":263957,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":0,"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":2,"date":"2016-01-06 11:03:16.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"all":[{"value":2,"date":"2016-01-06 08:52:03.000000000","_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":2,"date":"2016-01-06 10:36:27.000000000","_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-01-06 00:24:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"all":[{"value":0,"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":1,"date":"2016-01-06 10:36:27.000000000","_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-01-06 00:24:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-01-06 08:52:03.000000000","updated_by":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"reviewer":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"state":"REVIEWER"},{"updated":"2016-01-06 10:36:27.000000000","updated_by":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"reviewer":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"state":"REVIEWER"},{"updated":"2016-01-06 11:03:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8bba1b409dd45e102705f3fd1bc9dc3595b65260","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-01-05 23:53:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3f1b081494bfa40b982c42d181429028ed748b65","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-01-06 00:11:39.000000000","message":"Topic changed from PyMySQL-driver to puppet/liberty","accounts_in_message":[],"_revision_number":1},{"id":"da2fe6d788dcb236a1a9c00fbff9b77d220e2c0d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-01-06 00:24:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"281be86b380d1d3c7a7c6dd77c7a948b68fa659c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-01-06 00:24:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"937b49f2a1f4592e1a3b7a9c675b25c2996220d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-06 01:08:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/57/263957/1/check-tripleo/gate-tripleo-ci-f22-nonha/86f9d98/ : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1ffc953c6ca025365ad2a2f61498ed7ed452441d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-06 01:49:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-lint/a3250ca/ : SUCCESS in 48s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/ccc602e/ : SUCCESS in 3m 03s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/5b7f025/ : SUCCESS in 2m 58s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/d24c670/ : SUCCESS in 8m 21s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/aed759c/ : SUCCESS in 8m 00s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/9eae98a/ : SUCCESS in 10m 54s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/84166af/ : SUCCESS in 10m 47s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/49b1b4b/ : SUCCESS in 12m 09s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/8d09780/ : SUCCESS in 10m 09s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/aaa1052/ : SUCCESS in 10m 14s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/e691e7c/ : SUCCESS in 11m 54s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/28e44a3/ : SUCCESS in 11m 35s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/57/263957/1/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/a3eec55/ : SUCCESS in 28m 46s (non-voting)\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/7493b52/ : SUCCESS in 30m 18s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/57/263957/1/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/9c0e1bc/ : SUCCESS in 27m 09s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/57/263957/1/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/c6d149a/ : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c041337b2054026c08e45b1c909c26e0d3ed86e6","author":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"date":"2016-01-06 08:52:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58fa720ec2cf5685ec8ac8646cf04959b540d56e","author":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"date":"2016-01-06 10:36:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"570571dbc91df2b4d7fc4b70a49cb4fdd8f48ddd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-06 10:36:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"595b2b0d24474b3d3ded2030f90f575007c76531","date":"2016-01-06 11:03:16.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"c6c31a2a58849ca262a926b0b197c401e8deb952","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-06 11:03:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-lint/26953dc/ : SUCCESS in 51s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/b7e0e3c/ : SUCCESS in 2m 55s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/7eaa3e0/ : SUCCESS in 3m 01s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/6f6b3e3/ : SUCCESS in 8m 15s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/0bea30c/ : SUCCESS in 9m 17s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/34eea70/ : SUCCESS in 10m 53s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/a7904b7/ : SUCCESS in 13m 33s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/c141bcc/ : SUCCESS in 10m 48s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/7d9e884/ : SUCCESS in 8m 51s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/056f017/ : SUCCESS in 9m 45s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/57/263957/1/gate/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/b937347/ : SUCCESS in 10m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"972c4d6e76a539c59b2df78b25ca03859f2144bb","revisions":{"972c4d6e76a539c59b2df78b25ca03859f2144bb":{"kind":"REWORK","_number":1,"created":"2016-01-05 23:53:57.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/57/263957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/57/263957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/57/263957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/57/263957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/57/263957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/57/263957/1"}}},"commit":{"parents":[{"commit":"a697c9ecd4b86f579231316604c60ec39b9a6916","subject":"Make Keystone_endpoint match service by name/type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a697c9ecd4b86f579231316604c60ec39b9a6916"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2015-11-18 10:57:36.000000000","tz":180},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-01-05 23:53:57.000000000","tz":0},"subject":"Support of PyMySQL driver for MySQL backend","message":"Support of PyMySQL driver for MySQL backend\n\nAdd ability to use python-pymysql library\nas backend for MySQL connections.\nUpdate acceptance tests to use pyMySQL.\nDocs: https://wiki.openstack.org/wiki/PyMySQL_evaluation\nThe same implementation as it\u0027s done for keystone:\nhttps://review.openstack.org/#/c/242134/\n\nChange-Id: Ibc5b9717d681d9a090777d4fe64b906c3c07ec5d\n(cherry picked from commit f813a16a6999a300760ceb3939f90d9f683e6d69)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/972c4d6e76a539c59b2df78b25ca03859f2144bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/972c4d6e76a539c59b2df78b25ca03859f2144bb"}]},"parents_data":[{"branch_name":"refs/heads/stable/liberty","commit_id":"a697c9ecd4b86f579231316604c60ec39b9a6916","is_merged_in_target_branch":true,"change_id":"Idafa03acb82b87763d6f3dcfc2cbfd5b6a802ce0","change_number":260676,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/liberty"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
