)]}'
{"id":"openstack%2Fpuppet-nova~263959","triplet_id":"openstack%2Fpuppet-nova~stable%2Fliberty~I6ba84a2636411a0be76603d47ca09ee4c93fbfe6","project":"openstack/puppet-nova","branch":"stable/liberty","topic":"puppet/liberty","hashtags":[],"change_id":"I6ba84a2636411a0be76603d47ca09ee4c93fbfe6","subject":"Follow-up on PyMySQL support for Red Hat platforms","status":"MERGED","created":"2016-01-05 23:55:22.000000000","updated":"2016-02-25 01:33:50.000000000","submitted":"2016-02-25 01:33:49.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5867cfeaa99afc9ecd63b02d2a37a4fb8a07098a","_number":263959,"virtual_id_number":263959,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-02-25 01:33:49.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-02-24 15:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":2,"date":"2016-02-25 00:37:57.000000000","_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":1,"date":"2016-02-25 00:37:57.000000000","_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-01-05 23:55:22.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2016-02-24 15:13:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-02-25 00:37:57.000000000","updated_by":{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},"reviewer":{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},"state":"REVIEWER"},{"updated":"2016-02-25 01:33:49.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"2cda142cbdbf91470d5e7d32462c3dee09bed8e1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-01-05 23:55:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6eeb166ceecdbc55b57dee9d9a99c1ba4a3ae731","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-01-06 00:25:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"82c8ee554502d1250635ff13442db66146b7e2a2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-06 01:48:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-lint/5dc682e/ : SUCCESS in 35s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/c303ca7/ : SUCCESS in 3m 14s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/dec76cf/ : SUCCESS in 3m 00s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/a6470df/ : FAILURE in 7m 50s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/a9398b2/ : FAILURE in 8m 18s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/fe69cf1/ : FAILURE in 11m 48s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/187b672/ : FAILURE in 10m 54s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/d89d3d1/ : FAILURE in 13m 01s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/5601dff/ : FAILURE in 10m 50s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/6aaaa21/ : FAILURE in 10m 34s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/d3218d8/ : SUCCESS in 11m 48s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/a7d4de3/ : SUCCESS in 10m 19s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/59/263959/1/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/a4c32a2/ : SUCCESS in 26m 48s (non-voting)\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/d6f7700/ : SUCCESS in 27m 25s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/59/263959/1/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/18d0bf0/ : SUCCESS in 27m 53s (non-voting)\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/59/263959/1/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/c68dc2c/ : SUCCESS in 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1beeb3685bfceb235e187e8cf033586194ac2f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-06 02:41:50.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/59/263959/1/check-tripleo/gate-tripleo-ci-f22-nonha/d6b065f/ : FAILURE in 2h 46m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f3946770563f4b41af4c67b7a6c7017ee6adac79","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-01-06 21:30:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"e883d673d4b52300befbe0e7e59cd87b5d019203","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-02-24 12:41:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"6eada43f962a4a6e4b93da083d984fbe261fdc5b","author":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"date":"2016-02-24 13:02:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ccc23b8f1b833bc57f2c509787042a1f588b9506","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-24 13:38:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-lint/49fa5f4/ : SUCCESS in 45s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/0f855b6/ : SUCCESS in 4m 05s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/47a8e6a/ : SUCCESS in 3m 09s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/0e9261f/ : SUCCESS in 8m 39s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/48f4dfd/ : SUCCESS in 8m 52s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/dd2aef6/ : SUCCESS in 12m 08s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/d19faf3/ : SUCCESS in 8m 48s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/b49444f/ : SUCCESS in 15m 29s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/69e2594/ : SUCCESS in 13m 36s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/cd625a4/ : SUCCESS in 9m 58s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/78c415e/ : SUCCESS in 14m 17s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/c6c2153/ : SUCCESS in 10m 56s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/59/263959/2/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/e74ceb3/ : SUCCESS in 25m 20s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/4822f5c/ : SUCCESS in 22m 14s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/59/263959/2/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/7c6da41/ : SUCCESS in 27m 10s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/59/263959/2/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/6a8d18b/ : SUCCESS in 30m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3415d534b1dfdcd36901789ba2bee37da5fd8b12","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-02-24 15:13:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1343c750f88eda76b2bc7c8fa1f286aa9ea00efe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-24 15:49:23.000000000","message":"Patch Set 2:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/59/263959/2/check-tripleo/gate-tripleo-ci-f22-nonha/9e7d150/ : FAILURE in 2h 46m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2779831f5003ec86a3759715defc229cde1fe129","author":{"_account_id":7423,"name":"Cody Herriges","email":"cody@herriges.org","username":"ody"},"date":"2016-02-25 00:37:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd6d83698ce1031b23626c7d896f8396e79d4964","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-25 00:39:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"d2cd3af659b9504375bd8d9ba98bf7c0345d6a33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-25 01:33:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-lint/0c9c80f/ : SUCCESS in 1m 21s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/65356a2/ : SUCCESS in 3m 36s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/47004e4/ : SUCCESS in 3m 25s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/24fe839/ : SUCCESS in 6m 27s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/a6b5365/ : SUCCESS in 6m 56s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/217ab9f/ : SUCCESS in 13m 01s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/00fc2df/ : SUCCESS in 11m 30s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/d448603/ : SUCCESS in 15m 18s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/8a773cf/ : SUCCESS in 8m 34s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/742621b/ : SUCCESS in 12m 25s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/0ec4ca3/ : SUCCESS in 13m 21s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/59/263959/2/gate/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/0094ba4/ : SUCCESS in 32m 15s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/e4ff225/ : SUCCESS in 39m 26s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/59/263959/2/gate/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/4c76e9d/ : SUCCESS in 27m 06s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/59/263959/2/gate/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/1505495/ : SUCCESS in 23m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a580f32e52da6f76913e6e0e2199a4ad50262d89","date":"2016-02-25 01:33:50.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1fddb71670f986c730382d0f8103812587d4a90","revisions":{"d70fcee407b27fd1f0f7ceaf739caf095001623d":{"kind":"REWORK","_number":1,"created":"2016-01-05 23:55:22.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/59/263959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/59/263959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/1"}}},"commit":{"parents":[{"commit":"972c4d6e76a539c59b2df78b25ca03859f2144bb","subject":"Support of PyMySQL driver for MySQL backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/972c4d6e76a539c59b2df78b25ca03859f2144bb"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2015-12-02 13:23:25.000000000","tz":60},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-01-05 23:55:10.000000000","tz":-300},"subject":"Follow-up on PyMySQL support for Red Hat platforms","message":"Follow-up on PyMySQL support for Red Hat platforms\n\nRely on packaging dependencies to avoid issues caused by different\npackage names between Fedora and RHEL (python-PyMySQL vs\npython2-PyMySQL).\n\nhttps://review.openstack.org/#/c/245229/4/spec/classes/neutron_db_spec.rb\nincludes all the discussion that led to this.\n\nChange-Id: I6ba84a2636411a0be76603d47ca09ee4c93fbfe6\n(cherry picked from commit bd64e9274b33e89814a223240f3c954a71c8bca3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d70fcee407b27fd1f0f7ceaf739caf095001623d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d70fcee407b27fd1f0f7ceaf739caf095001623d"}]},"parents_data":[{"branch_name":"refs/heads/stable/liberty","commit_id":"972c4d6e76a539c59b2df78b25ca03859f2144bb","is_merged_in_target_branch":true,"change_id":"Ibc5b9717d681d9a090777d4fe64b906c3c07ec5d","change_number":263957,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/liberty"},"a1fddb71670f986c730382d0f8103812587d4a90":{"kind":"REWORK","_number":2,"created":"2016-02-24 13:02:29.000000000","uploader":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"ref":"refs/changes/59/263959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/59/263959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/59/263959/2"}}},"commit":{"parents":[{"commit":"a8849eb13662ec61bd311c70e63419bc984e6464","subject":"Supports IPv6 in vnc proxy.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a8849eb13662ec61bd311c70e63419bc984e6464"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2015-12-02 13:23:25.000000000","tz":60},"committer":{"name":"Lukas Bezdicka","email":"lbezdick@redhat.com","date":"2016-02-24 13:02:17.000000000","tz":60},"subject":"Follow-up on PyMySQL support for Red Hat platforms","message":"Follow-up on PyMySQL support for Red Hat platforms\n\nRely on packaging dependencies to avoid issues caused by different\npackage names between Fedora and RHEL (python-PyMySQL vs\npython2-PyMySQL).\n\nhttps://review.openstack.org/#/c/245229/4/spec/classes/neutron_db_spec.rb\nincludes all the discussion that led to this.\n\nChange-Id: I6ba84a2636411a0be76603d47ca09ee4c93fbfe6\n(cherry picked from commit bd64e9274b33e89814a223240f3c954a71c8bca3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a1fddb71670f986c730382d0f8103812587d4a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a1fddb71670f986c730382d0f8103812587d4a90"}]},"parents_data":[{"branch_name":"refs/heads/stable/liberty","commit_id":"a8849eb13662ec61bd311c70e63419bc984e6464","is_merged_in_target_branch":true,"change_id":"I12fb8f01b5bdca2bfea856c8a5e158fa1d87a15b","change_number":281346,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/stable/liberty"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
