)]}'
{"id":"openstack%2Fpuppet-nova~292309","triplet_id":"openstack%2Fpuppet-nova~master~I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8","project":"openstack/puppet-nova","branch":"master","hashtags":[],"change_id":"I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8","subject":"Use oslo module for messaging (rabbit) configuration","status":"MERGED","created":"2016-03-14 11:32:17.000000000","updated":"2016-04-11 23:50:28.000000000","submitted":"2016-04-11 23:50:28.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"317321fa46ce3ee242592e0afc68f5286f7da8dc","_number":292309,"virtual_id_number":292309,"owner":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":2,"date":"2016-04-11 23:50:28.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2016-04-10 18:40:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":1,"date":"2016-04-11 06:40:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2016-04-11 20:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2016-04-11 23:10:42.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-03-29 16:45:21.000000000","updated_by":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"reviewer":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"state":"REVIEWER"},{"updated":"2016-03-30 18:13:14.000000000","updated_by":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-11 06:40:00.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2016-04-11 20:44:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-04-11 23:10:42.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-04-11 23:50:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"346ad83c3dcf505b959a9efafdd0898f939bbd65","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 11:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cf3ae98e66c4b38b8f5030f5886b6872d37ce12","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 12:05:25.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a1339ec8613733d323166b8f587e929d9fa2bb79","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 12:07:02.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ca272a607e5638b078be6c320ae00b716fdf5eca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 12:07:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-lint/3b82389/ : SUCCESS in 3m 29s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/d2d3535/ : SUCCESS in 2m 54s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/8c3ead6/ : SUCCESS in 3m 30s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/777ccf3/ : SUCCESS in 6m 08s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/c720b98/ : SUCCESS in 9m 29s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/388b79e/ : SUCCESS in 14m 08s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/7147396/ : SUCCESS in 10m 28s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/bdaeb31/ : SUCCESS in 11m 57s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/a842e9a/ : SUCCESS in 10m 39s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/6a5f78d/ : SUCCESS in 13m 38s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/7e315d0/ : SUCCESS in 14m 28s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/be86c6b/ : SUCCESS in 14m 47s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/09/292309/1/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/7c85cb8/ : SUCCESS in 29m 53s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/888e2e0/ : SUCCESS in 33m 45s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/09/292309/1/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/0bc5066/ : SUCCESS in 26m 05s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/13defde/ : SUCCESS in 30m 15s\n- gate-puppet-openstack-integration-scenario003-tempest-dsvm-trusty http://logs.openstack.org/09/292309/1/check/gate-puppet-openstack-integration-scenario003-tempest-dsvm-trusty/4da5cab/ : SUCCESS in 22m 32s (non-voting)\n- gate-puppet-openstack-integration-scenario003-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/1/check/gate-puppet-openstack-integration-scenario003-tempest-dsvm-centos7/ad3dad2/ : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"21e3668f6ae898627dcc9f9f7e1d5b2a69a6aeb6","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-03-14 12:08:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d56aa1ce1d3e0c7a8570c9110296ea61a00068ac","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-14 12:37:40.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/428/ : SUCCESS\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/545/ : SUCCESS\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/516/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2ad0f2504faefe340ae4c26521e4a2ad9a74cd98","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 12:57:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/1/check-tripleo/gate-tripleo-ci-f22-nonha/b071f6d/ : SUCCESS in 1h 25m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ed33a3ec45af083f10ecbfefb6d67c9c26ae35da","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 13:03:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e4533deb364313ab9ef2c0df078f720d061475b","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 14:30:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9fb850a626b6b4b1ea5a6e82dd4f57b27bf062f2","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 17:07:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85639658e156dca02cd2bce8769675d3de4bc243","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 17:07:47.000000000","message":"Patch Set 2: Workflow-1\n\nIt won\u0027t pass CI without https://review.openstack.org/#/c/292378/","accounts_in_message":[],"_revision_number":2},{"id":"e347b017163a1ef63917d39ac0bc569785e315b1","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-14 17:11:41.000000000","message":"Topic add_kombu_compression removed","accounts_in_message":[],"_revision_number":2},{"id":"ca0bed9d0197630b30cf26fb7f891ac5c76615f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 17:22:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-lint/d07cf5a/ : SUCCESS in 4m 44s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/67e595d/ : SUCCESS in 3m 17s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/0b0514c/ : SUCCESS in 3m 42s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/a29fec3/ : FAILURE in 10m 38s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/b48b731/ : FAILURE in 9m 52s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/4a52642/ : FAILURE in 9m 52s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/b4317cf/ : FAILURE in 10m 17s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/e855113/ : FAILURE in 11m 52s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/4d1ce4b/ : FAILURE in 9m 33s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/8815a7a/ : FAILURE in 12m 51s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/1687d02/ : FAILURE in 4m 12s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/0a92d25/ : FAILURE in 5m 10s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/09/292309/2/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/175606e/ : FAILURE in 2m 18s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/171744c/ : FAILURE in 3m 08s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/09/292309/2/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/bb29c81/ : FAILURE in 2m 13s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/f5870da/ : FAILURE in 1m 53s\n- gate-puppet-openstack-integration-scenario003-tempest-dsvm-trusty http://logs.openstack.org/09/292309/2/check/gate-puppet-openstack-integration-scenario003-tempest-dsvm-trusty/01a8d53/ : FAILURE in 2m 13s (non-voting)\n- gate-puppet-openstack-integration-scenario003-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/2/check/gate-puppet-openstack-integration-scenario003-tempest-dsvm-centos7/257d8a4/ : FAILURE in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"cf354c80560bbe810dc876935b2a593e27a141f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 17:30:20.000000000","message":"Patch Set 2:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/2/check-tripleo/gate-tripleo-ci-f22-nonha/35ae0e5/ : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2192370731930707ccef898298e0811af9dd7173","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-14 18:12:46.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/454/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/574/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/541/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c56d760265177c9ab643ebb6d0d31cb5af24cbc2","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-15 15:14:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba3d67a9d88cc17d3b7a4e8c0f7d6d9f7c8972af","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-15 15:34:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-lint/fb06dfb/ : SUCCESS in 3m 15s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/077c768/ : SUCCESS in 2m 17s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/bc8f7a9/ : SUCCESS in 7m 33s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/76a1fff/ : FAILURE in 18m 34s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/76fa7f2/ : FAILURE in 13m 01s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/1b027cd/ : FAILURE in 10m 19s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/76b2e69/ : FAILURE in 10m 06s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/1cfb619/ : FAILURE in 10m 10s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/12ea639/ : FAILURE in 9m 04s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/683681d/ : FAILURE in 12m 07s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/505ffb7/ : FAILURE in 3m 52s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/c687673/ : FAILURE in 7m 16s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty http://logs.openstack.org/09/292309/3/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-trusty/79c3f1f/ : FAILURE in 2m 52s\n- gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-openstack-integration-scenario001-tempest-dsvm-centos7/37072ba/ : FAILURE in 3m 05s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty http://logs.openstack.org/09/292309/3/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-trusty/8b72c6c/ : FAILURE in 2m 08s\n- gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-openstack-integration-scenario002-tempest-dsvm-centos7/4a055dc/ : FAILURE in 2m 56s\n- gate-puppet-openstack-integration-scenario003-tempest-dsvm-trusty http://logs.openstack.org/09/292309/3/check/gate-puppet-openstack-integration-scenario003-tempest-dsvm-trusty/898a8c5/ : FAILURE in 3m 28s (non-voting)\n- gate-puppet-openstack-integration-scenario003-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/3/check/gate-puppet-openstack-integration-scenario003-tempest-dsvm-centos7/8247cbd/ : FAILURE in 1m 56s","accounts_in_message":[],"_revision_number":3},{"id":"dc2013d90efa1c5d56c8182850c749a89d975594","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-15 15:37:27.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/3/check-tripleo/gate-tripleo-ci-f22-nonha/2e185c2/ : FAILURE in 21m 29s","accounts_in_message":[],"_revision_number":3},{"id":"622ad1f8038754efcc41dffb09072c34a16e0e4b","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-15 16:42:47.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/473/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/593/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/560/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"e207aad0ade6d8bcf5c61acc8ef170096b79e093","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-15 19:10:30.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"851a6b2e081f91051bf01de2ae250d0a9fe1cda2","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-17 10:13:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"398ed9de1ebe71a295965b1db54e7871b0a5124f","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-17 10:14:05.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"9d0e4be0b742ce57c2452ebdf13feba57bce3d25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-17 10:34:43.000000000","message":"Patch Set 4:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/4/check-tripleo/gate-tripleo-ci-f22-nonha/a05772d/ : FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":4},{"id":"be8f1fbdabcf2e282b894580cc74e1242bf0323b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-17 10:48:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-lint/f80f500/ : SUCCESS in 1m 47s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/e2bccdd/ : SUCCESS in 4m 11s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/0e75ccd/ : SUCCESS in 3m 18s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/131b85c/ : SUCCESS in 8m 31s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/a848d9f/ : SUCCESS in 9m 57s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/2321c74/ : SUCCESS in 10m 47s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/d992f90/ : SUCCESS in 11m 27s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/4a527e7/ : SUCCESS in 11m 25s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/e7cd622/ : SUCCESS in 9m 25s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/b90ae77/ : SUCCESS in 17m 41s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/4a291ed/ : FAILURE in 9m 28s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/7133600/ : FAILURE in 14m 17s\n- gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-trusty http://logs.openstack.org/09/292309/4/check/gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-trusty/ce15e90/ : SUCCESS in 26m 37s\n- gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-centos7/bd18856/ : SUCCESS in 33m 13s\n- gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-trusty http://logs.openstack.org/09/292309/4/check/gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-trusty/9d86611/ : SUCCESS in 26m 16s\n- gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-centos7/9382dec/ : SUCCESS in 31m 58s\n- gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-trusty http://logs.openstack.org/09/292309/4/check/gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-trusty/da29e63/ : SUCCESS in 20m 19s\n- gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/4/check/gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-centos7/05de075/ : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":4},{"id":"bcb32bc8f9f414061445f21aacc884f144271268","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-17 11:07:03.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/516/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/637/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/602/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"cb4dd811464ee32ed01852437ba7fa9b24418139","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-17 11:34:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d18367f44d5d7d89b3612beccd4646da6384ef4e","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-17 11:34:11.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"18dd06b8d87a471437a07534145116e70c9b134f","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-17 11:46:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b87e7e643f0a88f4a82b572b8469471c0b4de15e","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-17 11:46:49.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/517/ : ABORTED\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/638/ : ABORTED\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/603/ : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"ccddda4616ed6ac5462c4a525936a8db1f01889d","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-17 11:49:24.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"fd33fa6ea5bb225d84e1a6a8a52078c69fc78685","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-17 12:10:27.000000000","message":"Patch Set 6:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/6/check-tripleo/gate-tripleo-ci-f22-nonha/499d1a3/ : FAILURE in 23m 24s","accounts_in_message":[],"_revision_number":6},{"id":"52470ee94d5de20ebe39d16fa6e8a83f72a72153","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-17 12:25:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-lint/07b088c/ : SUCCESS in 4m 02s\n- gate-puppet-nova-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-syntax-3-dsvm-centos7/a24388f/ : SUCCESS in 3m 33s\n- gate-puppet-nova-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-syntax-4-dsvm-centos7/74a19dc/ : SUCCESS in 2m 11s\n- gate-puppet-nova-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-unit-3.3-dsvm-centos7/061dfdc/ : SUCCESS in 6m 22s\n- gate-puppet-nova-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-unit-3.4-dsvm-centos7/4531af6/ : SUCCESS in 6m 51s\n- gate-puppet-nova-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-unit-3.6-dsvm-centos7/afbc5fb/ : SUCCESS in 13m 17s\n- gate-puppet-nova-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-unit-3.7-dsvm-centos7/75ee3c1/ : SUCCESS in 13m 54s\n- gate-puppet-nova-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-unit-3.8-dsvm-centos7/1dba09a/ : SUCCESS in 13m 43s\n- gate-puppet-nova-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-unit-4.0-dsvm-centos7/62c0e8a/ : SUCCESS in 9m 40s\n- gate-puppet-nova-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-unit-latest-dsvm-centos7/b5c6b28/ : SUCCESS in 10m 29s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-trusty/33dfdcf/ : SUCCESS in 12m 25s\n- gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-nova-puppet-beaker-rspec-dsvm-centos7/795c432/ : SUCCESS in 8m 54s\n- gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-trusty http://logs.openstack.org/09/292309/6/check/gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-trusty/80d47be/ : SUCCESS in 26m 33s\n- gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-openstack-integration-3-scenario001-tempest-dsvm-centos7/33f9ba7/ : SUCCESS in 36m 19s\n- gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-trusty http://logs.openstack.org/09/292309/6/check/gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-trusty/5322ac1/ : SUCCESS in 26m 15s\n- gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-openstack-integration-3-scenario002-tempest-dsvm-centos7/3f1c673/ : SUCCESS in 30m 45s\n- gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-trusty http://logs.openstack.org/09/292309/6/check/gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-trusty/28f0be2/ : SUCCESS in 21m 37s\n- gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-centos7 http://logs.openstack.org/09/292309/6/check/gate-puppet-openstack-integration-3-scenario003-tempest-dsvm-centos7/b0687e6/ : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":6},{"id":"d760b5c95b0dd238a18dd21f6b6313d05641789d","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-17 13:18:42.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/518/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/639/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/604/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"56f980aacbad5380f060bd1047e81bb7a3ed3c9a","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-29 13:07:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"40088630725bd5eb7aca5b1eb7dff1ea56371297","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-29 13:53:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-lint/8c350ee/ : SUCCESS in 2m 11s\n- gate-puppet-nova-puppet-syntax-3-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-syntax-3-devstack-centos7/8db3139/ : SUCCESS in 13m 42s\n- gate-puppet-nova-puppet-syntax-4-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-syntax-4-devstack-centos7/7439b11/ : SUCCESS in 2m 32s\n- gate-puppet-nova-puppet-unit-3.3-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.3-devstack-centos7/c3c0a49/ : SUCCESS in 13m 08s\n- gate-puppet-nova-puppet-unit-3.4-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.4-devstack-centos7/ba8e1df/ : SUCCESS in 9m 52s\n- gate-puppet-nova-puppet-unit-3.6-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.6-devstack-centos7/5cc4b5e/ : SUCCESS in 14m 46s\n- gate-puppet-nova-puppet-unit-3.7-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.7-devstack-centos7/5e46fb5/ : SUCCESS in 11m 48s\n- gate-puppet-nova-puppet-unit-3.8-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.8-devstack-centos7/48b9e72/ : SUCCESS in 20m 49s\n- gate-puppet-nova-puppet-unit-4.0-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-4.0-devstack-centos7/ae887d1/ : SUCCESS in 8m 11s\n- gate-puppet-nova-puppet-unit-latest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-latest-devstack-centos7/673bc9f/ : SUCCESS in 11m 10s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/03a2adf/ : SUCCESS in 11m 53s\n- gate-puppet-nova-puppet-beaker-rspec-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-beaker-rspec-devstack-centos7/09d6685/ : SUCCESS in 14m 30s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/b0b32ed/ : SUCCESS in 44m 41s\n- gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7/874e1de/ : SUCCESS in 34m 51s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/ff0e988/ : SUCCESS in 33m 26s\n- gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7/3f360ca/ : SUCCESS in 32m 56s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/13a6b0e/ : SUCCESS in 31m 52s\n- gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7/a215f4e/ : SUCCESS in 28m 15s","accounts_in_message":[],"_revision_number":7},{"id":"4bb613e58e978499cb4143695d57585a9b230782","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-29 13:59:53.000000000","message":"Patch Set 7:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/7/check-tripleo/gate-tripleo-ci-f22-nonha/f03a24e/ : FAILURE in 29m 18s","accounts_in_message":[],"_revision_number":7},{"id":"df267c8258edb9f1e6501793d66a04afdd2e15cf","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-03-29 14:17:05.000000000","message":"Patch Set 7: Code-Review-1\n\nexcellent work! we\u0027re adding the module to tripleo today, also please add a release note.\nWe\u0027ll do a recheck afterwards.","accounts_in_message":[],"_revision_number":7},{"id":"70a9ce1c9bdf351864c9abb07511879eeceb1986","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-03-29 14:26:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6778718fbc2689077d9b9abce00af4a17bbd52b5","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-29 15:56:27.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/866/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/988/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/946/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"a9c965d13dea7eb38e0b78a4c5c41540255986c7","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-03-29 16:45:21.000000000","message":"Patch Set 7: Code-Review-1\n\nAdd please puppet-oslo to dependency list: https://github.com/openstack/puppet-nova/blob/master/metadata.json#L33","accounts_in_message":[],"_revision_number":7},{"id":"d80b90ec51dcce42378e3f8ec655c64edb9852c1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-03-29 18:06:28.000000000","message":"Patch Set 7:\n\nrecheck tripleo should have puppet-oslo now","accounts_in_message":[],"_revision_number":7},{"id":"a6879d6a4306e9432bd912ea4d2b27ca7b2335bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-29 18:54:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-lint/7a5835e/ : SUCCESS in 1m 51s\n- gate-puppet-nova-puppet-syntax-3-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-syntax-3-devstack-centos7/2d22992/ : SUCCESS in 3m 53s\n- gate-puppet-nova-puppet-syntax-4-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-syntax-4-devstack-centos7/0104ccf/ : SUCCESS in 14m 45s\n- gate-puppet-nova-puppet-unit-3.3-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.3-devstack-centos7/24bbb7e/ : SUCCESS in 8m 24s\n- gate-puppet-nova-puppet-unit-3.4-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.4-devstack-centos7/6f9c922/ : SUCCESS in 13m 10s\n- gate-puppet-nova-puppet-unit-3.6-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.6-devstack-centos7/c3f27d3/ : SUCCESS in 13m 58s\n- gate-puppet-nova-puppet-unit-3.7-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.7-devstack-centos7/be30e60/ : SUCCESS in 16m 03s\n- gate-puppet-nova-puppet-unit-3.8-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-3.8-devstack-centos7/db89cf0/ : SUCCESS in 15m 09s\n- gate-puppet-nova-puppet-unit-4.0-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-4.0-devstack-centos7/3ea864f/ : SUCCESS in 13m 28s\n- gate-puppet-nova-puppet-unit-latest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-unit-latest-devstack-centos7/31e6604/ : SUCCESS in 14m 41s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/d06bbab/ : SUCCESS in 20m 10s\n- gate-puppet-nova-puppet-beaker-rspec-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-nova-puppet-beaker-rspec-devstack-centos7/88dd2ff/ : SUCCESS in 16m 48s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/d9c585c/ : SUCCESS in 18m 54s\n- gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7/d1698bf/ : SUCCESS in 32m 08s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/87629c0/ : SUCCESS in 22m 13s\n- gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7/62e3f71/ : SUCCESS in 34m 21s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/8ffda10/ : SUCCESS in 22m 58s\n- gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7 http://logs.openstack.org/09/292309/7/check/gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7/925108a/ : SUCCESS in 26m 22s","accounts_in_message":[],"_revision_number":7},{"id":"f251187bac1b99b34d4cb27409909a368fd9479e","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-29 19:54:00.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/878/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/1002/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/958/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"7905037c145dc932ef6f4dca28ae0aa6139361bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-29 21:03:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/7/check-tripleo/gate-tripleo-ci-f22-nonha/c449777/ : SUCCESS in 1h 53m 13s","accounts_in_message":[],"_revision_number":7},{"id":"77afe86f554ba7b7ff390861bc789e70c93823b9","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-30 11:36:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b663db42fbe3fb284aa6f322c6fbeda9ac5156f6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 11:51:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-lint/55c0e2b/ : SUCCESS in 2m 36s\n- gate-puppet-nova-puppet-syntax-3-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-syntax-3-devstack-centos7/c3dba05/ : SUCCESS in 4m 21s\n- gate-puppet-nova-puppet-syntax-4-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-syntax-4-devstack-centos7/faffb55/ : SUCCESS in 2m 23s\n- gate-puppet-nova-puppet-unit-3.3-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-unit-3.3-devstack-centos7/bfe4020/ : FAILURE in 6m 42s\n- gate-puppet-nova-puppet-unit-3.4-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-unit-3.4-devstack-centos7/f23e9a1/ : FAILURE in 6m 54s\n- gate-puppet-nova-puppet-unit-3.6-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-unit-3.6-devstack-centos7/626f637/ : FAILURE in 12m 08s\n- gate-puppet-nova-puppet-unit-3.7-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-unit-3.7-devstack-centos7/b0b1e3d/ : FAILURE in 8m 57s\n- gate-puppet-nova-puppet-unit-3.8-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-unit-3.8-devstack-centos7/34bb5fc/ : FAILURE in 10m 44s\n- gate-puppet-nova-puppet-unit-4.0-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-unit-4.0-devstack-centos7/872404e/ : FAILURE in 8m 11s\n- gate-puppet-nova-puppet-unit-latest-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-unit-latest-devstack-centos7/3a08885/ : FAILURE in 10m 42s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/cdc0ffa/ : FAILURE in 4m 01s\n- gate-puppet-nova-puppet-beaker-rspec-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-nova-puppet-beaker-rspec-devstack-centos7/692c90f/ : FAILURE in 6m 04s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/8/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/f3d4636/ : FAILURE in 3m 28s\n- gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7/361aebd/ : FAILURE in 3m 40s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/8/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/4210bae/ : FAILURE in 3m 20s\n- gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7/1a90fb4/ : FAILURE in 3m 23s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/8/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/dd761ed/ : FAILURE in 2m 56s\n- gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7 http://logs.openstack.org/09/292309/8/check/gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7/c9e91a7/ : FAILURE in 3m 30s\n- gate-puppet-nova-releasenotes http://docs-draft.openstack.org/09/292309/8/check/gate-puppet-nova-releasenotes/d914e09//releasenotes/build/html/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":8},{"id":"285c3e57e47ccbdfa7c0d05a79006394964f0ffb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 12:02:09.000000000","message":"Patch Set 8:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/8/check-tripleo/gate-tripleo-ci-f22-nonha/c44ff9d/ : FAILURE in 24m 29s","accounts_in_message":[],"_revision_number":8},{"id":"3d7811061f84b97cb97acc67881393b74be6e163","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-30 12:09:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"85c4dfe8e36442fa78791435eef10059b8cf9f92","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-30 12:09:39.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/1057/ : ABORTED\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/1011/ : ABORTED\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/931/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"fc672ac34e4fed1c455a9060f1bba6174911bde8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 12:33:06.000000000","message":"Patch Set 9:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/9/check-tripleo/gate-tripleo-ci-f22-nonha/39f9eaf/ : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":9},{"id":"3dda51128769af6ab8b607f229ebff1cd8f69ca2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 12:50:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-lint/94fdb91/ : SUCCESS in 2m 13s\n- gate-puppet-nova-puppet-syntax-3-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-syntax-3-devstack-centos7/0b3fcf8/ : SUCCESS in 3m 28s\n- gate-puppet-nova-puppet-syntax-4-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-syntax-4-devstack-centos7/c6da0f0/ : SUCCESS in 3m 18s\n- gate-puppet-nova-puppet-unit-3.3-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.3-devstack-centos7/9ad20be/ : SUCCESS in 9m 58s\n- gate-puppet-nova-puppet-unit-3.4-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.4-devstack-centos7/ae2db8a/ : SUCCESS in 7m 03s\n- gate-puppet-nova-puppet-unit-3.6-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.6-devstack-centos7/3187979/ : SUCCESS in 9m 17s\n- gate-puppet-nova-puppet-unit-3.7-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.7-devstack-centos7/8ed7105/ : SUCCESS in 9m 47s\n- gate-puppet-nova-puppet-unit-3.8-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.8-devstack-centos7/dc2735b/ : SUCCESS in 17m 38s\n- gate-puppet-nova-puppet-unit-4.0-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-4.0-devstack-centos7/17f0f90/ : SUCCESS in 11m 44s\n- gate-puppet-nova-puppet-unit-latest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-latest-devstack-centos7/ce7899e/ : SUCCESS in 16m 15s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/1f82a5d/ : SUCCESS in 7m 15s\n- gate-puppet-nova-puppet-beaker-rspec-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-beaker-rspec-devstack-centos7/0667f16/ : SUCCESS in 15m 12s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/1868b4b/ : SUCCESS in 30m 43s\n- gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7/f91a30d/ : SUCCESS in 34m 21s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/aa2f83b/ : SUCCESS in 19m 10s\n- gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7/02c39aa/ : SUCCESS in 33m 06s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/faebd98/ : SUCCESS in 15m 54s\n- gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7/3866c2b/ : SUCCESS in 18m 44s\n- gate-puppet-nova-releasenotes http://docs-draft.openstack.org/09/292309/9/check/gate-puppet-nova-releasenotes/5f4f403//releasenotes/build/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":9},{"id":"dceb5348bb9bd8a24d2722d70728b747e2133ca6","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-30 13:23:14.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/934/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/1062/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/1014/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"4ec98389a12d6c01eefa1a29c3e9aff227b5c77f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-03-30 13:29:15.000000000","message":"Patch Set 9:\n\nrecheck we need tripleo CI \u0026 fuel CI working","accounts_in_message":[],"_revision_number":9},{"id":"244b5a1ff2c71c5d63564a57de585ae4cad1c421","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 13:57:08.000000000","message":"Patch Set 9:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/9/check-tripleo/gate-tripleo-ci-f22-nonha/71b891e/ : FAILURE in 20m 29s","accounts_in_message":[],"_revision_number":9},{"id":"a6ce85d82ab8b757a0a58aa033dd5ba7df3c4d38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 14:18:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-lint/bb0346d/ : SUCCESS in 1m 40s\n- gate-puppet-nova-puppet-syntax-3-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-syntax-3-devstack-centos7/626b319/ : SUCCESS in 2m 40s\n- gate-puppet-nova-puppet-syntax-4-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-syntax-4-devstack-centos7/28f6fce/ : SUCCESS in 2m 35s\n- gate-puppet-nova-puppet-unit-3.3-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.3-devstack-centos7/05b109b/ : SUCCESS in 6m 28s\n- gate-puppet-nova-puppet-unit-3.4-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.4-devstack-centos7/1669a5a/ : SUCCESS in 7m 05s\n- gate-puppet-nova-puppet-unit-3.6-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.6-devstack-centos7/8c25d1f/ : SUCCESS in 10m 05s\n- gate-puppet-nova-puppet-unit-3.7-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.7-devstack-centos7/ee4b23b/ : SUCCESS in 16m 13s\n- gate-puppet-nova-puppet-unit-3.8-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-3.8-devstack-centos7/d3c5369/ : SUCCESS in 11m 39s\n- gate-puppet-nova-puppet-unit-4.0-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-4.0-devstack-centos7/429c63d/ : SUCCESS in 9m 39s\n- gate-puppet-nova-puppet-unit-latest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-unit-latest-devstack-centos7/53cd9a9/ : SUCCESS in 10m 52s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/b9249e7/ : SUCCESS in 11m 58s\n- gate-puppet-nova-puppet-beaker-rspec-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-nova-puppet-beaker-rspec-devstack-centos7/21a6de6/ : SUCCESS in 15m 29s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/7b81fc7/ : SUCCESS in 31m 23s\n- gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario001-tempest-devstack-centos7/a3d0df0/ : SUCCESS in 41m 53s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/85b760e/ : SUCCESS in 19m 11s\n- gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario002-tempest-devstack-centos7/e705a04/ : SUCCESS in 39m 34s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/015356f/ : SUCCESS in 24m 42s\n- gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7 http://logs.openstack.org/09/292309/9/check/gate-puppet-openstack-integration-3-scenario003-tempest-devstack-centos7/6d530a0/ : SUCCESS in 28m 02s\n- gate-puppet-nova-releasenotes http://docs-draft.openstack.org/09/292309/9/check/gate-puppet-nova-releasenotes/ee1ea92//releasenotes/build/html/ : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":9},{"id":"9eb40931f580ac1ae776ada2b8f4d23cdc3180e8","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-03-30 14:25:51.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"30a7a02810c60753bd9e6a0b0a8c64b16eaba68c","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-03-30 18:13:14.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/944/ : FAILURE\n\n* puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/puppet-openstack.fuel.noop/1073/ : FAILURE\n\n* puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/1024/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"418f40432b2096aac54aeb25e6ba2f60db19e7c5","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-04-07 15:19:17.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"2668a944116efe45b7d6abe1dd9e74c41885a34e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-07 16:19:38.000000000","message":"Patch Set 10:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/10/check-tripleo/gate-tripleo-ci-f22-nonha/c7b9866/ : FAILURE in 45m 42s","accounts_in_message":[],"_revision_number":10},{"id":"d02191b9e6f565fca18c30d2241de331e04dd838","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-07 17:18:16.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-lint/85bda1d/ : SUCCESS in 5m 07s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-syntax-3-centos-7/b37ee25/ : SUCCESS in 5m 19s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-syntax-4-centos-7/08d71b3/ : SUCCESS in 5m 44s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.3-centos-7/7847435/ : SUCCESS in 8m 44s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.4-centos-7/1aff2ab/ : SUCCESS in 9m 43s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.6-centos-7/cb71c4c/ : SUCCESS in 12m 53s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.7-centos-7/6b29a42/ : SUCCESS in 11m 13s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.8-centos-7/8d8fcbb/ : SUCCESS in 16m 22s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-4.0-centos-7/3de78ac/ : SUCCESS in 12m 26s\n- gate-puppet-nova-puppet-unit-latest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-latest-centos-7/b05c7b8/ : SUCCESS in 12m 57s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/ec4db50/ : SUCCESS in 7m 24s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-beaker-rspec-centos-7/f6014c0/ : SUCCESS in 13m 40s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/c56d160/ : SUCCESS in 19m 32s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/7db4f9a/ : SUCCESS in 39m 51s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/bd80c41/ : SUCCESS in 31m 08s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/adcee90/ : FAILURE in 1h 00m 33s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/f1bf484/ : SUCCESS in 17m 45s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/00fe3b2/ : SUCCESS in 35m 48s\n- gate-puppet-nova-releasenotes http://docs-draft.openstack.org/09/292309/10/check/gate-puppet-nova-releasenotes/b977c21//releasenotes/build/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":10},{"id":"39a824e4cc9b7afb1096878a8536d149171a42ee","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-04-07 19:31:20.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/65/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/66/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/68/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"71021eb9fae0a4a9b215a6c16fbb4cd8b1c9ae06","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-04-07 21:11:30.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"ed16bd3212344f26451d1a2ded93d679cd26c751","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-07 22:16:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-lint/24461f4/ : SUCCESS in 7m 41s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-syntax-3-centos-7/ed64033/ : SUCCESS in 7m 07s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-syntax-4-centos-7/dbeacaf/ : SUCCESS in 6m 59s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.3-centos-7/f8d0e49/ : SUCCESS in 6m 13s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.4-centos-7/b85657d/ : SUCCESS in 11m 26s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.6-centos-7/674a909/ : SUCCESS in 10m 03s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.7-centos-7/34d637d/ : SUCCESS in 19m 59s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.8-centos-7/342a1f1/ : SUCCESS in 13m 10s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-4.0-centos-7/c830dcb/ : SUCCESS in 10m 29s\n- gate-puppet-nova-puppet-unit-latest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-latest-centos-7/4f29628/ : SUCCESS in 13m 29s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/548eaa7/ : SUCCESS in 11m 23s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-beaker-rspec-centos-7/1f04858/ : SUCCESS in 13m 25s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/df031ed/ : SUCCESS in 28m 54s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/dafb793/ : SUCCESS in 34m 51s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/803bd1a/ : SUCCESS in 21m 13s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/c9c03c1/ : SUCCESS in 35m 41s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/baf42cb/ : SUCCESS in 17m 14s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/d41af1e/ : SUCCESS in 30m 10s\n- gate-puppet-nova-releasenotes http://docs-draft.openstack.org/09/292309/10/check/gate-puppet-nova-releasenotes/126e9d2//releasenotes/build/html/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":10},{"id":"1a01102b56bb02c2399aa68e52135ba77f6084f3","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-04-07 22:27:15.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/77/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/78/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/82/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"0c0978d6eb788eb65c0ddebafab9e95c53b17a61","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-08 00:14:00.000000000","message":"Patch Set 10:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/10/check-tripleo/gate-tripleo-ci-f22-nonha/94c87a4/ : FAILURE in 1h 37m 02s","accounts_in_message":[],"_revision_number":10},{"id":"723b1e2415dbd37f3f3c0217c92d759e255bde89","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-04-08 14:24:40.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"eaeff763148fb97a7319cc9b098ea2164bf3a0b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-08 15:04:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-lint/981adac/ : SUCCESS in 4m 16s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-syntax-3-centos-7/deb788c/ : SUCCESS in 3m 15s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-syntax-4-centos-7/34f8ec3/ : SUCCESS in 4m 07s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.3-centos-7/c42ae28/ : SUCCESS in 7m 27s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.4-centos-7/2e011a2/ : SUCCESS in 10m 30s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.6-centos-7/5d9fe41/ : SUCCESS in 13m 11s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.7-centos-7/d6c5b0b/ : SUCCESS in 10m 55s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-3.8-centos-7/af82e95/ : SUCCESS in 12m 58s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-4.0-centos-7/4f87824/ : SUCCESS in 8m 19s\n- gate-puppet-nova-puppet-unit-latest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-unit-latest-centos-7/d5fc40b/ : SUCCESS in 23m 08s (non-voting)\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/8402fc5/ : SUCCESS in 8m 12s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-nova-puppet-beaker-rspec-centos-7/a208727/ : SUCCESS in 16m 59s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/e21b150/ : SUCCESS in 21m 57s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/0ba494b/ : SUCCESS in 32m 29s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/f501d48/ : SUCCESS in 30m 06s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/d65b71e/ : SUCCESS in 33m 17s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/eb96c72/ : SUCCESS in 21m 01s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/09/292309/10/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/57677bc/ : SUCCESS in 31m 37s\n- gate-puppet-nova-releasenotes http://docs-draft.openstack.org/09/292309/10/check/gate-puppet-nova-releasenotes/0e74d32//releasenotes/build/html/ : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":10},{"id":"283f0ac7f9d70f1fc9e0d143ab8c3d9d0be5cfe7","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-04-08 16:00:32.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/88/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/89/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/93/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"10d8d6d4aa4d0a45dd2d6dd16d110af867feecd5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-08 16:26:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/09/292309/10/check-tripleo/gate-tripleo-ci-f22-nonha/fa1e93a/ : SUCCESS in 1h 57m 01s","accounts_in_message":[],"_revision_number":10},{"id":"a2a9aba5faa1b8a65257fad503dbef666367e7d5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-04-10 18:40:38.000000000","message":"Patch Set 10: Code-Review+2\n\nExcellent work here!","accounts_in_message":[],"_revision_number":10},{"id":"24fdd78d367638fcfd3054dbb69a8b00f1953c8d","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2016-04-11 06:40:00.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"b0d3af74c777655942bf3dda61e5c86105250343","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-04-11 20:44:28.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7d6ebc6bf44d91857fad9530e9fbdeba7a8cf773","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-04-11 23:10:42.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"d8a6db7bb2819630b46993e8bb1f73959652b5e1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-11 23:10:47.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":10},{"id":"4d8d9750e532790b9b44e359ff6bfd0a0b50d602","date":"2016-04-11 23:50:28.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":10},{"id":"317321fa46ce3ee242592e0afc68f5286f7da8dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-11 23:50:28.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-lint/04c34f4/ : SUCCESS in 4m 07s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-syntax-3-centos-7/9e20d4c/ : SUCCESS in 3m 52s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-syntax-4-centos-7/c17a563/ : SUCCESS in 3m 09s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-unit-3.3-centos-7/72f57a0/ : SUCCESS in 7m 08s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-unit-3.4-centos-7/d40920d/ : SUCCESS in 7m 34s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-unit-3.6-centos-7/c451877/ : SUCCESS in 10m 41s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-unit-3.7-centos-7/57aed66/ : SUCCESS in 16m 15s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-unit-3.8-centos-7/eec270b/ : SUCCESS in 13m 33s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-unit-4.0-centos-7/649806e/ : SUCCESS in 10m 33s\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/4c2b55c/ : SUCCESS in 12m 27s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-nova-puppet-beaker-rspec-centos-7/efbb9b2/ : SUCCESS in 13m 12s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/gate/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/0cec54c/ : SUCCESS in 24m 50s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/b95b2f1/ : SUCCESS in 32m 58s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/gate/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/eb6b6b1/ : SUCCESS in 32m 36s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/6c3892c/ : SUCCESS in 34m 32s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty http://logs.openstack.org/09/292309/10/gate/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-trusty/3e43986/ : SUCCESS in 21m 51s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/09/292309/10/gate/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/a76a2b8/ : SUCCESS in 38m 24s\n- gate-puppet-nova-releasenotes http://docs-draft.openstack.org/09/292309/10/gate/gate-puppet-nova-releasenotes/e3accda//releasenotes/build/html/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"94c3bbed7112aad97ae634844c275d74d5b3afed","revisions":{"a23d38ec1bf02884824c18d206bb17a4bb52f4f9":{"kind":"REWORK","_number":1,"created":"2016-03-14 11:32:17.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/1"}}},"commit":{"parents":[{"commit":"7ccbdec3f85b1a02ab14c3997f2ef15d956a2869","subject":"Revert \"cpu_allocation_ratio / ram_allocation_ratio options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/7ccbdec3f85b1a02ab14c3997f2ef15d956a2869"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:32:11.000000000","tz":180},"subject":"Add kombu_compression parameter for Rabbit","message":"Add kombu_compression parameter for Rabbit\n\nOslo.messaging message compression was implemented\nin Mitaka here https://review.openstack.org/#/c/249215\n\nThe option should be disabled by default because it is experimental\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a23d38ec1bf02884824c18d206bb17a4bb52f4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a23d38ec1bf02884824c18d206bb17a4bb52f4f9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7ccbdec3f85b1a02ab14c3997f2ef15d956a2869","is_merged_in_target_branch":true,"change_id":"I5e9e21ff4b49618843cc3d4376c5f323aafd79cf","change_number":291665,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a53d97806d003c2aec1046252ab379b9c9955174":{"kind":"REWORK","_number":2,"created":"2016-03-14 17:07:16.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/2"}}},"commit":{"parents":[{"commit":"40a62aba4182a9a418a9ab59b325a93cff1072bc","subject":"Add basic structure for ReNo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/40a62aba4182a9a418a9ab59b325a93cff1072bc"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 17:06:56.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuraion","message":"Use oslo module for messaging (rabbit) configuraion\n\nDepends on: Id41a07ad6a77e956493aca34fc892a51221d0aa2\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n\n  * update unit tests\n\n  * fix values for boolean variables\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a53d97806d003c2aec1046252ab379b9c9955174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a53d97806d003c2aec1046252ab379b9c9955174"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40a62aba4182a9a418a9ab59b325a93cff1072bc","is_merged_in_target_branch":true,"change_id":"Ia1e1159c9b86660c328680ef6519297d0921b5cc","change_number":292164,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"471ca860bdb6024d7f9e7eb9e2cc0757aa406187":{"kind":"REWORK","_number":3,"created":"2016-03-15 15:14:32.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/3"}}},"commit":{"parents":[{"commit":"40a62aba4182a9a418a9ab59b325a93cff1072bc","subject":"Add basic structure for ReNo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/40a62aba4182a9a418a9ab59b325a93cff1072bc"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-15 15:14:00.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuraion","message":"Use oslo module for messaging (rabbit) configuraion\n\nDepends on: Id41a07ad6a77e956493aca34fc892a51221d0aa2\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n\n  * update unit tests\n\n  * fix values for boolean variables\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/471ca860bdb6024d7f9e7eb9e2cc0757aa406187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/471ca860bdb6024d7f9e7eb9e2cc0757aa406187"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40a62aba4182a9a418a9ab59b325a93cff1072bc","is_merged_in_target_branch":true,"change_id":"Ia1e1159c9b86660c328680ef6519297d0921b5cc","change_number":292164,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8c8d37ce902db7a35d89a32c237f2b6fe8b81e93":{"kind":"REWORK","_number":4,"created":"2016-03-17 10:13:51.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/4"}}},"commit":{"parents":[{"commit":"40a62aba4182a9a418a9ab59b325a93cff1072bc","subject":"Add basic structure for ReNo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/40a62aba4182a9a418a9ab59b325a93cff1072bc"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-17 10:12:46.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuraion","message":"Use oslo module for messaging (rabbit) configuraion\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n  * update unit tests\n  * fix values for boolean variables\n  * specify type for each parameter\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8c8d37ce902db7a35d89a32c237f2b6fe8b81e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8c8d37ce902db7a35d89a32c237f2b6fe8b81e93"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40a62aba4182a9a418a9ab59b325a93cff1072bc","is_merged_in_target_branch":true,"change_id":"Ia1e1159c9b86660c328680ef6519297d0921b5cc","change_number":292164,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e5c705d28168ac50a90700e69500fc81c8d96345":{"kind":"REWORK","_number":5,"created":"2016-03-17 11:34:03.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/5"}}},"commit":{"parents":[{"commit":"40a62aba4182a9a418a9ab59b325a93cff1072bc","subject":"Add basic structure for ReNo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/40a62aba4182a9a418a9ab59b325a93cff1072bc"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-17 11:33:49.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuraion","message":"Use oslo module for messaging (rabbit) configuraion\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n  * update unit tests\n  * fix values for boolean variables\n  * specify type for each parameter\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/e5c705d28168ac50a90700e69500fc81c8d96345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/e5c705d28168ac50a90700e69500fc81c8d96345"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40a62aba4182a9a418a9ab59b325a93cff1072bc","is_merged_in_target_branch":true,"change_id":"Ia1e1159c9b86660c328680ef6519297d0921b5cc","change_number":292164,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8b1f5a135fc9fa101a4bbb5fd482431bf423510f":{"kind":"REWORK","_number":6,"created":"2016-03-17 11:46:47.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/6"}}},"commit":{"parents":[{"commit":"40a62aba4182a9a418a9ab59b325a93cff1072bc","subject":"Add basic structure for ReNo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/40a62aba4182a9a418a9ab59b325a93cff1072bc"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-17 11:46:34.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuraion","message":"Use oslo module for messaging (rabbit) configuraion\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n  * update unit tests\n  * fix values for boolean variables\n  * specify type for each parameter\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8b1f5a135fc9fa101a4bbb5fd482431bf423510f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8b1f5a135fc9fa101a4bbb5fd482431bf423510f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40a62aba4182a9a418a9ab59b325a93cff1072bc","is_merged_in_target_branch":true,"change_id":"Ia1e1159c9b86660c328680ef6519297d0921b5cc","change_number":292164,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bf7d4f2f08e554e4228c9c024e8346f5e669a202":{"kind":"REWORK","_number":7,"created":"2016-03-29 13:07:11.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/7"}}},"commit":{"parents":[{"commit":"9aa5ee10241dd66585be685427c178c2f05e051a","subject":"Autorequire nova security group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/9aa5ee10241dd66585be685427c178c2f05e051a"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-29 13:00:11.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuraion","message":"Use oslo module for messaging (rabbit) configuraion\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n  * update unit tests\n  * fix values for boolean variables\n  * specify type for each parameter\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/bf7d4f2f08e554e4228c9c024e8346f5e669a202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/bf7d4f2f08e554e4228c9c024e8346f5e669a202"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9aa5ee10241dd66585be685427c178c2f05e051a","is_merged_in_target_branch":true,"change_id":"I89aa036c8a24d919c15ceb909f0c567dad3c6032","change_number":296548,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"149da4ca66aa29e8be16cee57cb1963b22a38a7a":{"kind":"REWORK","_number":8,"created":"2016-03-30 11:36:13.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/8"}}},"commit":{"parents":[{"commit":"2c9bb31867778ea32cb779b7efdb86ae69a4059e","subject":"Merge \"Use OpenstackClient for nova providers auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/2c9bb31867778ea32cb779b7efdb86ae69a4059e"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-30 11:35:58.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuration","message":"Use oslo module for messaging (rabbit) configuration\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n  * update unit tests\n  * fix values for boolean variables\n  * specify type for each parameter\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/149da4ca66aa29e8be16cee57cb1963b22a38a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/149da4ca66aa29e8be16cee57cb1963b22a38a7a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2c9bb31867778ea32cb779b7efdb86ae69a4059e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fec467a12751965986e097f77a958d957abcc42a":{"kind":"REWORK","_number":9,"created":"2016-03-30 12:09:37.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/9"}}},"commit":{"parents":[{"commit":"2c9bb31867778ea32cb779b7efdb86ae69a4059e","subject":"Merge \"Use OpenstackClient for nova providers auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/2c9bb31867778ea32cb779b7efdb86ae69a4059e"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-30 12:09:25.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuration","message":"Use oslo module for messaging (rabbit) configuration\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n  * update unit tests\n  * fix values for boolean variables\n  * specify type for each parameter\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/fec467a12751965986e097f77a958d957abcc42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/fec467a12751965986e097f77a958d957abcc42a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2c9bb31867778ea32cb779b7efdb86ae69a4059e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"94c3bbed7112aad97ae634844c275d74d5b3afed":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2016-04-07 15:19:17.000000000","uploader":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"ref":"refs/changes/09/292309/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/09/292309/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/09/292309/10"}}},"commit":{"parents":[{"commit":"f0709b7d23cee998d6901aa91ab9fd120a8371af","subject":"Merge \"Gemfile: rely on puppet-openstack_spec_helper for dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f0709b7d23cee998d6901aa91ab9fd120a8371af"}]}],"author":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-03-14 11:29:59.000000000","tz":180},"committer":{"name":"iberezovskiy","email":"iberezovskiy@mirantis.com","date":"2016-04-07 11:04:14.000000000","tz":180},"subject":"Use oslo module for messaging (rabbit) configuration","message":"Use oslo module for messaging (rabbit) configuration\n\nAdditonal changes:\n\n  * switch messaging related parameters to $::os_service_default\n    as they are used in oslo::messaging_rabbit define\n  * update unit tests\n  * fix values for boolean variables\n  * specify type for each parameter\n\nChange-Id: I9d03db24fafdd1f5fd2141cdc599b3ab212a04d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/94c3bbed7112aad97ae634844c275d74d5b3afed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/94c3bbed7112aad97ae634844c275d74d5b3afed"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f0709b7d23cee998d6901aa91ab9fd120a8371af","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
