)]}'
{"id":"openstack%2Fpuppet-nova~325000","triplet_id":"openstack%2Fpuppet-nova~master~I4ae83b6c3049f5fe9030588a878bed8d6d3ac9c2","project":"openstack/puppet-nova","branch":"master","topic":"change/wiki","hashtags":[],"change_id":"I4ae83b6c3049f5fe9030588a878bed8d6d3ac9c2","subject":"Change wiki to docs","status":"MERGED","created":"2016-06-03 02:24:46.000000000","updated":"2016-06-03 16:08:09.000000000","submitted":"2016-06-03 16:08:09.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3c29a316d2e672b4c9f25d338defce437d56a425","_number":325000,"virtual_id_number":325000,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2016-06-03 16:08:09.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"all":[{"value":2,"date":"2016-06-03 08:58:21.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":2,"date":"2016-06-03 13:53:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":1,"date":"2016-06-03 13:53:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-03 08:58:21.000000000","updated_by":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"reviewer":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"state":"REVIEWER"},{"updated":"2016-06-03 13:53:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-06-03 16:08:09.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8ecf82cbd31827714a41754c61b49a8014f79cfb","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2016-06-03 02:24:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f91ae8eb17a3c8d55fdf24c85bb3710153fbfbfd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 02:30:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/00/325000/1/check/gate-puppet-nova-puppet-lint/a8a2d55/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"470d9c7529aa0a56c80904849626457b3c70033c","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-06-03 03:45:48.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/1966/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/1953/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/2000/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8918c3f52d684c65ad07b30ffdb5f6d58998cc7f","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-06-03 08:58:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8d2e0619c40a0ae99b3b1b86769d1748dcb516d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-06-03 13:53:36.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aaddafd6215581afc8896a716b986fd051012e5a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 13:54:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"f32f15465f863252936e88b603ded4088e318948","date":"2016-06-03 16:08:09.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"f5db730cce5c2d7ea7044b45b24a993fb7111512","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 16:08:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/00/325000/1/gate/gate-puppet-nova-puppet-lint/197bd9c/ : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a00406c4e78dd33978057592d92383d92e6a10fa","revisions":{"a00406c4e78dd33978057592d92383d92e6a10fa":{"kind":"REWORK","_number":1,"created":"2016-06-03 02:24:46.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/00/325000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/00/325000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/00/325000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/00/325000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/00/325000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/00/325000/1"}}},"commit":{"parents":[{"commit":"1055fbbd195dab6d956db2bc2fd057972de895dd","subject":"Merge \"Ensure nova-network service is up before creating networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/1055fbbd195dab6d956db2bc2fd057972de895dd"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2016-06-03 01:55:37.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2016-06-03 02:24:36.000000000","tz":-180},"subject":"Change wiki to docs","message":"Change wiki to docs\n\nChange-Id: I4ae83b6c3049f5fe9030588a878bed8d6d3ac9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a00406c4e78dd33978057592d92383d92e6a10fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a00406c4e78dd33978057592d92383d92e6a10fa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1055fbbd195dab6d956db2bc2fd057972de895dd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
