)]}'
{"id":"openstack%2Fpuppet-nova~325106","triplet_id":"openstack%2Fpuppet-nova~stable%2Fliberty~Iaf21b70f274d2c5769c6bb117697efe9e74873fd","project":"openstack/puppet-nova","branch":"stable/liberty","topic":"novanet","hashtags":[],"change_id":"Iaf21b70f274d2c5769c6bb117697efe9e74873fd","subject":"Ensure nova-network service is up before creating networks","status":"MERGED","created":"2016-06-03 07:55:35.000000000","updated":"2016-06-06 22:37:39.000000000","submitted":"2016-06-06 22:37:39.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6da3d65b1fd151cd53069ca6d06af9f0b65682b1","_number":325106,"virtual_id_number":325106,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2016-06-06 22:37:39.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":18795,"name":"mkarpin","email":"mkarpin@mirantis.com","username":"nkarpin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2016-06-04 04:08:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-06-06 20:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2016-06-06 07:55:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18795,"name":"mkarpin","email":"mkarpin@mirantis.com","username":"nkarpin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-06-06 20:35:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":18795,"name":"mkarpin","email":"mkarpin@mirantis.com","username":"nkarpin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18795,"name":"mkarpin","email":"mkarpin@mirantis.com","username":"nkarpin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-04 04:08:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-06-06 07:55:24.000000000","updated_by":{"_account_id":18795,"name":"mkarpin","email":"mkarpin@mirantis.com","username":"nkarpin"},"reviewer":{"_account_id":18795,"name":"mkarpin","email":"mkarpin@mirantis.com","username":"nkarpin"},"state":"REVIEWER"},{"updated":"2016-06-06 20:35:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-06-06 22:37:39.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"321af809a0a4a93d77c7fb5dd6d3151b5a2bd520","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2016-06-03 07:55:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e29017015a7f2dc16af5ee9b480cdb814192fd6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 08:27:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-lint/2594a06/ : SUCCESS in 5m 23s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-3-centos-7/1b2ff68/ : SUCCESS in 4m 08s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-4-centos-7/5ec2c59/ : SUCCESS in 4m 56s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.3-centos-7/4b4e016/ : SUCCESS in 9m 02s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.4-centos-7/c126a48/ : SUCCESS in 8m 32s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.6-centos-7/e2a09aa/ : SUCCESS in 13m 26s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.7-centos-7/de9d1cf/ : SUCCESS in 12m 42s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.8-centos-7/a54994f/ : SUCCESS in 12m 54s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-4.0-centos-7/52a8bd3/ : SUCCESS in 12m 00s\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/97527e2/ : SUCCESS in 13m 19s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-centos-7/9afaa07/ : SUCCESS in 10m 29s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/b4dfe34/ : SUCCESS in 17m 19s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/93e2beb/ : UNSTABLE in 23m 54s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/aa74834/ : UNSTABLE in 26m 47s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/238c3d1/ : UNSTABLE in 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"de4beec5355db5b858f6b878941ef91257e8cdb4","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2016-06-03 10:00:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9eb7350ceade1efd9b5f60e37031d35b48b83f02","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 10:31:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-lint/26cef40/ : SUCCESS in 4m 22s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-3-centos-7/feebe8e/ : SUCCESS in 5m 36s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-4-centos-7/1de514e/ : SUCCESS in 4m 34s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.3-centos-7/c3a9c11/ : SUCCESS in 8m 16s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.4-centos-7/dd561a0/ : SUCCESS in 7m 53s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.6-centos-7/64404df/ : SUCCESS in 10m 37s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.7-centos-7/15656ad/ : SUCCESS in 13m 00s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.8-centos-7/c562ec4/ : SUCCESS in 13m 21s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-4.0-centos-7/9f5cd20/ : SUCCESS in 11m 28s\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/f3afdbc/ : UNSTABLE in 10m 31s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-centos-7/bf33b1d/ : UNSTABLE in 11m 59s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/e65e531/ : UNSTABLE in 15m 36s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/dfda0a4/ : UNSTABLE in 23m 29s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/8cbcaf8/ : UNSTABLE in 15m 33s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/9745ab1/ : UNSTABLE in 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fd8e42b56d5c118f3b6adb4328b972a466a60369","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 10:58:19.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/06/325106/1/check-tripleo/gate-tripleo-ci-f22-nonha/81524d3/ : FAILURE in 2h 44m 37s","accounts_in_message":[],"_revision_number":1},{"id":"161474abd1445a7977325c9e4cf57a1407f6e9d2","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2016-06-03 14:03:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e64e0897183c8a4ca4ca85e071ee8ae63d9d15e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 14:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-lint/f6c3818/ : SUCCESS in 6m 00s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-3-centos-7/db7404f/ : SUCCESS in 5m 01s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-4-centos-7/05b94c3/ : SUCCESS in 5m 55s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.3-centos-7/9b4754a/ : SUCCESS in 9m 32s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.4-centos-7/d31575c/ : SUCCESS in 10m 00s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.6-centos-7/ac31623/ : SUCCESS in 11m 08s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.7-centos-7/ad793b2/ : SUCCESS in 9m 52s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.8-centos-7/a8f55dd/ : SUCCESS in 9m 07s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-4.0-centos-7/5c413d2/ : SUCCESS in 9m 55s\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/ecd9c9f/ : SUCCESS in 14m 05s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-centos-7/cc292a7/ : SUCCESS in 11m 31s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/f57acc9/ : SUCCESS in 22m 22s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/f911c7a/ : SUCCESS in 26m 08s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/36b61ff/ : SUCCESS in 24m 41s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/13ba151/ : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d030eec8f9b9a74054563d3c55a40bcce861b93a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 16:01:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/06/325106/1/check-tripleo/gate-tripleo-ci-f22-nonha/c5085d5/ : SUCCESS in 1h 48m 16s","accounts_in_message":[],"_revision_number":1},{"id":"afa0e1cd297b29065b4fbc11a302caecb5e1aeda","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-06-04 04:08:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"231410ada5a7be6b975468b0b0b696cc7f1469dc","author":{"_account_id":18795,"name":"mkarpin","email":"mkarpin@mirantis.com","username":"nkarpin"},"date":"2016-06-06 07:55:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"526dc26e09136646211749287031f6e904187160","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-06-06 20:35:39.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2be93c61df7b93919309199f0caf36e5565aa4c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 20:36:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"478fcbb65a51e23a955ee728eb47c61b61ee0d44","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 21:39:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-lint/821d2c5/ : SUCCESS in 6m 15s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-syntax-3-centos-7/b76418f/ : SUCCESS in 3m 53s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-syntax-4-centos-7/f9960f0/ : SUCCESS in 2m 49s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.3-centos-7/19c210e/ : SUCCESS in 6m 25s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.4-centos-7/da691f3/ : SUCCESS in 7m 18s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.6-centos-7/2c2f2e1/ : SUCCESS in 10m 37s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.7-centos-7/d794381/ : SUCCESS in 10m 45s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.8-centos-7/d2a0c7f/ : SUCCESS in 10m 38s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-4.0-centos-7/c27af7c/ : SUCCESS in 11m 25s\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/e993578/ : SUCCESS in 14m 00s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-beaker-rspec-centos-7/f18bb1c/ : SUCCESS in 11m 42s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/c625a52/ : SUCCESS in 26m 17s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/9149568/ : FAILURE in 1h 00m 19s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/21a78bf/ : SUCCESS in 22m 52s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/fd25df8/ : SUCCESS in 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"34e02829df9fac248942219380498c20581ad80c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-06-06 21:40:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdafc6d29d9d316ea75606da3b6b6f6369801059","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 22:09:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-lint/25f0b6e/ : SUCCESS in 4m 48s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-3-centos-7/f018a28/ : SUCCESS in 3m 54s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-syntax-4-centos-7/7498483/ : SUCCESS in 4m 12s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.3-centos-7/d070d6e/ : SUCCESS in 8m 28s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.4-centos-7/90040e6/ : SUCCESS in 9m 02s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.6-centos-7/f23fc5c/ : SUCCESS in 11m 03s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.7-centos-7/d9f3a12/ : SUCCESS in 8m 11s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-3.8-centos-7/33e6606/ : SUCCESS in 11m 08s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-unit-4.0-centos-7/bc8191d/ : SUCCESS in 9m 35s\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/040ef02/ : SUCCESS in 11m 22s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-nova-puppet-beaker-rspec-centos-7/591539e/ : SUCCESS in 11m 23s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/9a9b97f/ : SUCCESS in 22m 51s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/347ebbc/ : SUCCESS in 21m 52s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/1f037fa/ : SUCCESS in 16m 01s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/06/325106/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/3395f50/ : SUCCESS in 24m 21s","accounts_in_message":[],"_revision_number":1},{"id":"518a2bd43bf83626041ab9a8a7cee1d0d8e30885","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 22:09:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"4cf6ed9d6e526da4007c454a521dd9873a2d43bd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 22:28:10.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-nonha http://logs.openstack.org/06/325106/1/check-tripleo/gate-tripleo-ci-centos-7-nonha/9f7cec5/ : FAILURE in 43m 05s","accounts_in_message":[],"_revision_number":1},{"id":"23f85723d4567d0d0fc76560893bee45569481d8","date":"2016-06-06 22:37:39.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"4158509d0efc258089bb14f6aa00e00e695f076e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 22:37:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-nova-puppet-lint http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-lint/2cfed69/ : SUCCESS in 4m 07s\n- gate-puppet-nova-puppet-syntax-3-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-syntax-3-centos-7/a1ba1e2/ : SUCCESS in 3m 55s\n- gate-puppet-nova-puppet-syntax-4-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-syntax-4-centos-7/4bc662e/ : SUCCESS in 4m 18s\n- gate-puppet-nova-puppet-unit-3.3-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.3-centos-7/3e74025/ : SUCCESS in 5m 52s\n- gate-puppet-nova-puppet-unit-3.4-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.4-centos-7/219829f/ : SUCCESS in 8m 05s\n- gate-puppet-nova-puppet-unit-3.6-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.6-centos-7/22858d7/ : SUCCESS in 11m 04s\n- gate-puppet-nova-puppet-unit-3.7-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.7-centos-7/8e22a45/ : SUCCESS in 9m 51s\n- gate-puppet-nova-puppet-unit-3.8-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-3.8-centos-7/8a9c7c5/ : SUCCESS in 11m 06s\n- gate-puppet-nova-puppet-unit-4.0-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-unit-4.0-centos-7/3781b62/ : SUCCESS in 10m 30s\n- gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-beaker-rspec-ubuntu-trusty/6912d3a/ : SUCCESS in 12m 38s\n- gate-puppet-nova-puppet-beaker-rspec-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-nova-puppet-beaker-rspec-centos-7/0e13483/ : SUCCESS in 10m 53s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-trusty/c866ea1/ : SUCCESS in 16m 46s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/ed2e259/ : SUCCESS in 22m 38s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-trusty/d5dba85/ : SUCCESS in 23m 16s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/06/325106/1/gate/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/db3ebb5/ : SUCCESS in 23m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c3f3f3eba9e5ec2b45058872504a278f6abca7f","revisions":{"6c3f3f3eba9e5ec2b45058872504a278f6abca7f":{"kind":"REWORK","_number":1,"created":"2016-06-03 07:55:35.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/06/325106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/06/325106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/06/325106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/06/325106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/06/325106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/06/325106/1"}}},"commit":{"parents":[{"commit":"806464321b6bf1e22a342e89d41ce631f3f8065b","subject":"Prepare 7.1.0 release (liberty)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/806464321b6bf1e22a342e89d41ce631f3f8065b"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2016-05-18 15:53:04.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2016-06-03 07:55:35.000000000","tz":0},"subject":"Ensure nova-network service is up before creating networks","message":"Ensure nova-network service is up before creating networks\n\nhttps://review.openstack.org/229895 made sure nova networks could\nonly be created if the nova-network service was enabled. However,\nit is still possible that Puppet tries to create them before the\nservice is started.\n\nThis patch should be backported to stable/mitaka and stable/liberty.\n\nChange-Id: Iaf21b70f274d2c5769c6bb117697efe9e74873fd\n(cherry picked from commit cd3611beb9801f2b91a63251810f1468d5d7e9a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/6c3f3f3eba9e5ec2b45058872504a278f6abca7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/6c3f3f3eba9e5ec2b45058872504a278f6abca7f"}]},"parents_data":[{"branch_name":"refs/heads/stable/liberty","commit_id":"806464321b6bf1e22a342e89d41ce631f3f8065b","is_merged_in_target_branch":true,"change_id":"I54730327bb35a403f8e866f51f98467332954971","change_number":319367,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/liberty"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
