)]}'
{"id":"openstack%2Fpuppet-nova~552658","triplet_id":"openstack%2Fpuppet-nova~master~Ieeb5aac0c70dd1b87b27379cd61cc52d9767c508","project":"openstack/puppet-nova","branch":"master","topic":"add_dns_entries_1534070_1","attention_set":{},"removed_from_attention_set":{"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-04-18 07:32:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ieeb5aac0c70dd1b87b27379cd61cc52d9767c508","subject":"Add call to retrieve DNS keytab","status":"ABANDONED","created":"2018-03-13 18:41:44.000000000","updated":"2022-04-18 07:32:53.000000000","total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ca434b3d373950299020852fbf3ebe42a425aa9a","_number":552658,"virtual_id_number":552658,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-05 22:24:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27999,"name":"lishaoyang","email":"lisy_email@163.com","username":"lisy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":-1,"date":"2021-01-06 04:39:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27999,"name":"lishaoyang","email":"lisy_email@163.com","username":"lisy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27999,"name":"lishaoyang","email":"lisy_email@163.com","username":"lisy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27999,"name":"lishaoyang","email":"lisy_email@163.com","username":"lisy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-16 21:30:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2018-03-19 08:29:28.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2018-03-22 09:52:47.000000000","updated_by":{"_account_id":27999,"name":"lishaoyang","email":"lisy_email@163.com","username":"lisy"},"reviewer":{"_account_id":27999,"name":"lishaoyang","email":"lisy_email@163.com","username":"lisy"},"state":"REVIEWER"},{"updated":"2018-10-26 05:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-29 12:56:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2018-11-06 15:50:02.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-01-06 04:39:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"eeccaf9e18dd54cec94337407ec471934d087b00","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2018-03-13 18:41:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1be5f1d71f173ef15184adcb1cfe303121dad3fe","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2018-03-13 18:47:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"86db4c150a6436a0dde7dc6e22dcc422833df0b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-13 21:22:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/58/552658/2/check/build-openstack-releasenotes/fd6a22a/html/ : SUCCESS in 2m 33s\n- puppet-openstack-lint http://logs.openstack.org/58/552658/2/check/puppet-openstack-lint/5a997c3/ : SUCCESS in 3m 29s\n- puppet-openstack-syntax-4 http://logs.openstack.org/58/552658/2/check/puppet-openstack-syntax-4/0c61b24/ : SUCCESS in 3m 05s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-unit-4.8-centos-7/c62deee/ : SUCCESS in 23m 51s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-unit-latest-centos-7/688c050/ : SUCCESS in 21m 18s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/73e615d/ : SUCCESS in 1h 02m 12s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/fa89cbf/ : SUCCESS in 1h 03m 01s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/9cbdd1a/ : SUCCESS in 56m 53s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/15d4aae/ : SUCCESS in 46m 46s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/107d1f1/ : SUCCESS in 41m 33s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/48edb21/ : SUCCESS in 43m 06s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/ad39507/ : SUCCESS in 36m 57s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/533ab5b/ : SUCCESS in 32m 10s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-beaker-centos-7/9fb71db/ : SUCCESS in 24m 52s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-beaker-xenial/4e314f1/ : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f835c7371b9328d3f36c65ab0843523dcb0a20d6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-03-16 21:30:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f49bfdc23b9a1122732467ba18834fd97e51e3ea","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-03-19 08:29:28.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b0ba4813e2b1fc4b8e72f150cf7065e3a06d0fd5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-03-21 02:50:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0dd0d91b56b7d635e28dd187221d0d7673d494fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-21 04:11:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-releasenotes http://logs.openstack.org/58/552658/2/check/build-openstack-releasenotes/324b3b5/html/ : SUCCESS in 2m 54s\n- puppet-openstack-lint http://logs.openstack.org/58/552658/2/check/puppet-openstack-lint/7d28bd2/ : SUCCESS in 3m 53s\n- puppet-openstack-syntax-4 http://logs.openstack.org/58/552658/2/check/puppet-openstack-syntax-4/29163e7/ : SUCCESS in 3m 39s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-unit-4.8-centos-7/8041bc8/ : SUCCESS in 20m 50s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-unit-latest-centos-7/82eeda4/ : SUCCESS in 19m 02s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/e1e05de/ : FAILURE in 5m 06s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/66929a1/ : FAILURE in 4m 42s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/5cf0bde/ : FAILURE in 6m 25s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/1c39255/ : FAILURE in 5m 45s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/dbb6e22/ : SUCCESS in 39m 12s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/0c2ea65/ : SUCCESS in 1h 02m 32s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/fc769b1/ : SUCCESS in 40m 55s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/0da3b70/ : SUCCESS in 31m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-beaker-centos-7/b1f1f3b/ : SUCCESS in 26m 32s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-beaker-xenial/026b0c5/ : SUCCESS in 20m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a12842c9053916648d23c502b827ca20a767d291","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-03-22 07:21:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"17ea9a16cb288a8195a40ff9fda4ffc3555a7e2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-22 08:29:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/58/552658/2/check/build-openstack-releasenotes/642c941/html/ : SUCCESS in 2m 50s\n- puppet-openstack-lint http://logs.openstack.org/58/552658/2/check/puppet-openstack-lint/835a247/ : SUCCESS in 3m 38s\n- puppet-openstack-syntax-4 http://logs.openstack.org/58/552658/2/check/puppet-openstack-syntax-4/6576068/ : SUCCESS in 3m 06s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-unit-4.8-centos-7/2638d9a/ : SUCCESS in 22m 49s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-unit-latest-centos-7/2bb1998/ : SUCCESS in 17m 05s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/664641f/ : SUCCESS in 1h 00m 58s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/4f2c95d/ : SUCCESS in 56m 57s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/4c024f4/ : SUCCESS in 1h 00m 33s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/8534eba/ : SUCCESS in 45m 56s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/c5b8c0f/ : SUCCESS in 40m 57s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/67cf4c5/ : SUCCESS in 39m 55s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/0a12c90/ : SUCCESS in 31m 56s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/304541c/ : SUCCESS in 32m 05s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/58/552658/2/check/puppet-openstack-beaker-centos-7/ce4e4f5/ : SUCCESS in 22m 53s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/58/552658/2/check/puppet-openstack-beaker-xenial/72b328b/ : SUCCESS in 15m 06s","accounts_in_message":[],"_revision_number":2},{"id":"22594748e39c8e54a4522c8f2567c13215ef7e33","author":{"_account_id":27999,"name":"lishaoyang","email":"lisy_email@163.com","username":"lisy"},"date":"2018-03-22 09:52:47.000000000","message":"Patch Set 2: Code-Review+1\n\nLgtm","accounts_in_message":[],"_revision_number":2},{"id":"a5ae4cc0749f7f41f98ef86a74423250f5d8cf86","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-03-23 01:45:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1d70473e3a9fd30a3b6987ab0c70892f2762d95b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-03-23 01:45:45.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a8093101c34d68c9eaef04e65e36ff9e9ed8c074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 02:56:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/58/552658/3/check/build-openstack-releasenotes/0c80157/html/ : SUCCESS in 2m 18s\n- puppet-openstack-lint http://logs.openstack.org/58/552658/3/check/puppet-openstack-lint/3f6656a/ : SUCCESS in 3m 40s\n- puppet-openstack-syntax-4 http://logs.openstack.org/58/552658/3/check/puppet-openstack-syntax-4/9e22dfc/ : SUCCESS in 3m 45s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/58/552658/3/check/puppet-openstack-unit-4.8-centos-7/6b93af5/ : SUCCESS in 23m 02s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/58/552658/3/check/puppet-openstack-unit-latest-centos-7/fcf7a34/ : SUCCESS in 20m 14s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/f5848f1/ : SUCCESS in 59m 45s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/00d08f2/ : SUCCESS in 1h 08m 12s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/df011aa/ : SUCCESS in 1h 05m 07s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/cc7e46e/ : SUCCESS in 49m 23s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/373ec40/ : SUCCESS in 42m 46s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/60e826f/ : SUCCESS in 41m 15s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/d269a1c/ : SUCCESS in 34m 01s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/58/552658/3/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/c234a02/ : SUCCESS in 33m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/58/552658/3/check/puppet-openstack-beaker-centos-7/743149b/ : SUCCESS in 25m 32s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/58/552658/3/check/puppet-openstack-beaker-xenial/d8cddc7/ : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":3},{"id":"f8d977d3ffd818ad81c594268c26384402f8e401","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-10-25 23:46:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b857d386589a2227276864a06b7523725f485f70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 05:23:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/58/552658/4/check/puppet-openstack-lint/34c8632/ : SUCCESS in 6m 24s\n- puppet-openstack-syntax-4 http://logs.openstack.org/58/552658/4/check/puppet-openstack-syntax-4/43a9879/ : SUCCESS in 3m 54s\n- puppet-openstack-syntax-5 http://logs.openstack.org/58/552658/4/check/puppet-openstack-syntax-5/118565d/ : SUCCESS in 3m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/58/552658/4/check/puppet-openstack-unit-4.8-centos-7/ad4b22c/ : SUCCESS in 21m 40s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/58/552658/4/check/puppet-openstack-unit-5.5-centos-7/117664a/ : SUCCESS in 20m 52s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/58/552658/4/check/puppet-openstack-unit-latest-centos-7/121b3eb/ : SUCCESS in 22m 56s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/e647062/ : SUCCESS in 48m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/2562129/ : SUCCESS in 1h 00m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/79b6113/ : SUCCESS in 1h 09m 40s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/6fd5d5f/ : SUCCESS in 1h 14m 36s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/1ae57c1/ : SUCCESS in 52m 50s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/92a2246/ : SUCCESS in 1h 04m 09s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/22a5196/ : SUCCESS in 50m 11s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/26d288b/ : SUCCESS in 54m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/4d59417/ : SUCCESS in 37m 02s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/6c04bb9/ : FAILURE in 47m 08s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/1b0c1bb/ : FAILURE in 38m 49s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/bc53506/ : FAILURE in 1h 07m 09s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/58/552658/4/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/9f00b83/ : TIMED_OUT in 1h 30m 47s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/58/552658/4/check/puppet-openstack-beaker-centos-7/682396c/ : SUCCESS in 24m 38s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/58/552658/4/check/puppet-openstack-beaker-ubuntu-bionic/b458505/ : SUCCESS in 36m 40s\n- tripleo-puppet-ci-centos-7-standalone http://logs.openstack.org/58/552658/4/check/tripleo-puppet-ci-centos-7-standalone/af12e87/ : SUCCESS in 1h 23m 55s\n- build-openstack-releasenotes http://logs.openstack.org/58/552658/4/check/build-openstack-releasenotes/e7039ec/html/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":4},{"id":"35c38938701499e556aa548067f315b8b3cb4f2d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-10-29 12:56:43.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"75311a1179d2ce7b2bbf674af851b2dc9c9e6a42","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-11-06 15:50:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0b53d5af00381562f2d627fa522202468125b476","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 15:39:57.000000000","message":"Patch Set 4:\n\n@alee, is this ever going to get merged in novajoin?","accounts_in_message":[],"_revision_number":4},{"id":"223a8bcaa6d2a424a965a2ef6a98090d1800a48f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-01-05 19:24:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"41ca12c2833a6d6a6fbe05b5b9004241682484a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 22:24:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21d1ffd217de4246854427061e28a6be : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1a704f4c06c4e5bbcaaf8fb5b92d04a : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1fc89c6562404668bf905c95ac7ed661 : FAILURE in 20m 18s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2041501363f044d3a29f3e008090cc09 : FAILURE in 17m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4979327babc43bfb943f9e3f7d13cc6 : FAILURE in 20m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee6ff8a71b40465299ccc5e5979d53df : SUCCESS in 55m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06f11d41d0e34c3bb277772af4e6299e : SUCCESS in 1h 05m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c661eab4be524549a0e2e18373d1c4f8 : SUCCESS in 59m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cd8bbff7fd546c29a9fa6fd7810565f : SUCCESS in 38m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/56a7f726eb9346289c31dc364cf52c7d : SUCCESS in 1h 25m 20s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f6b888fc5f6d401fa0e0d5f1feea89fc : SUCCESS in 1h 33m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8f6cb592dc41464e83a4a825fac724e2 : SUCCESS in 1h 49m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0cc4c204c28e403090cf27a3928443ae : SUCCESS in 1h 09m 38s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/68082c6ed1e14b6b93f0036f8e6c7839 : SUCCESS in 26m 38s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82907d9e93cf456eaba9e8d012e3a737 : SUCCESS in 26m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ef15a52091488fbdb36c386d385f20 : SUCCESS in 3m 57s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bc6b30a96646442a8e94a87dcef3b40f : SUCCESS in 2h 53m 09s","accounts_in_message":[],"_revision_number":5},{"id":"90123b5d8b17a884fdd9445b48dace2eb8c56e96","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-06 04:39:55.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nIMHO the current default values are inappropriate. They are too specific to TripleO and we should use more generic defaults in puppet-nova.\nYou can override these values in puppet-tripleo or tripleo-heat-templates.\n\nAlso, please consider to file a bug in launchpad in addition to bugzilla, to follow how we track the bug/issue in openstack projects.","accounts_in_message":[],"_revision_number":5},{"id":"09de8991155cdc36df932ee7d122a8aee5949782","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-18 07:32:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ca434b3d373950299020852fbf3ebe42a425aa9a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-18 07:32:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"934618462cc7a571a5c1c17ea502155b2227195d","revisions":{"9dd5c47581573b77485e8a10d994d65854462560":{"kind":"REWORK","_number":1,"created":"2018-03-13 18:41:44.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/58/552658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/58/552658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/1"}}},"commit":{"parents":[{"commit":"45b3c8cfa1f8ed4ab80204d9c0e3fb9af95fb306","subject":"Add \u0027openstack-db\u0027 tag to db-sync Exec resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/45b3c8cfa1f8ed4ab80204d9c0e3fb9af95fb306"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2018-03-07 23:03:40.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2018-03-13 18:18:26.000000000","tz":-240},"subject":"Add call to retrieve DNS keytab","message":"Add call to retrieve DNS keytab\n\nPartially resolves BZ 1534070\nChange-Id: Ieeb5aac0c70dd1b87b27379cd61cc52d9767c508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/9dd5c47581573b77485e8a10d994d65854462560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/9dd5c47581573b77485e8a10d994d65854462560"}]},"branch":"refs/heads/master"},"91727c4749717a01848ed6372f9cabba194914a1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-03-13 18:47:25.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/58/552658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/58/552658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/2"}}},"commit":{"parents":[{"commit":"45b3c8cfa1f8ed4ab80204d9c0e3fb9af95fb306","subject":"Add \u0027openstack-db\u0027 tag to db-sync Exec resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/45b3c8cfa1f8ed4ab80204d9c0e3fb9af95fb306"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2018-03-07 23:03:40.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2018-03-13 18:42:23.000000000","tz":-240},"subject":"Add call to retrieve DNS keytab","message":"Add call to retrieve DNS keytab\n\nPartially resolves BZ 1534070\nDepends-On: I589e896f5eda4bfa22f42babe590e4ead3870236\nChange-Id: Ieeb5aac0c70dd1b87b27379cd61cc52d9767c508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/91727c4749717a01848ed6372f9cabba194914a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/91727c4749717a01848ed6372f9cabba194914a1"}]},"branch":"refs/heads/master"},"12d5cc1e6b366261a41bb0fe8e84d5c90d080385":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-03-23 01:45:35.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/58/552658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/58/552658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/3"}}},"commit":{"parents":[{"commit":"8d3fe77ce044bbfc51272dd46866427cba54d5d0","subject":"Merge \"oslo middleware: check puppet resource instead of actual config in spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8d3fe77ce044bbfc51272dd46866427cba54d5d0"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2018-03-07 23:03:40.000000000","tz":-300},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2018-03-23 01:45:35.000000000","tz":0},"subject":"Add call to retrieve DNS keytab","message":"Add call to retrieve DNS keytab\n\nPartially resolves BZ 1534070\nDepends-On: I589e896f5eda4bfa22f42babe590e4ead3870236\nChange-Id: Ieeb5aac0c70dd1b87b27379cd61cc52d9767c508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/12d5cc1e6b366261a41bb0fe8e84d5c90d080385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/12d5cc1e6b366261a41bb0fe8e84d5c90d080385"}]},"branch":"refs/heads/master"},"f089c3374f703a30a1183e18243d388d4b08cdeb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-10-25 23:46:44.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/58/552658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/58/552658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/4"}}},"commit":{"parents":[{"commit":"3d877926f09d9c0c0f331025256c4b55214c1269","subject":"Fixup functions for 5.5.7+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/3d877926f09d9c0c0f331025256c4b55214c1269"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2018-03-07 23:03:40.000000000","tz":-300},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-10-25 23:46:44.000000000","tz":0},"subject":"Add call to retrieve DNS keytab","message":"Add call to retrieve DNS keytab\n\nPartially resolves BZ 1534070\nDepends-On: I589e896f5eda4bfa22f42babe590e4ead3870236\nChange-Id: Ieeb5aac0c70dd1b87b27379cd61cc52d9767c508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f089c3374f703a30a1183e18243d388d4b08cdeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f089c3374f703a30a1183e18243d388d4b08cdeb"}]},"branch":"refs/heads/master"},"934618462cc7a571a5c1c17ea502155b2227195d":{"kind":"REWORK","_number":5,"created":"2021-01-05 19:24:42.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/58/552658/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/58/552658/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/58/552658/5"}}},"commit":{"parents":[{"commit":"39d10a9a71dbbffa6394fd2bb32e7cdd5ccd7e64","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/39d10a9a71dbbffa6394fd2bb32e7cdd5ccd7e64"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2018-03-07 23:03:40.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-01-05 19:23:47.000000000","tz":-300},"subject":"Add call to retrieve DNS keytab","message":"Add call to retrieve DNS keytab\n\nPartially resolves BZ 1534070\nDepends-On: I589e896f5eda4bfa22f42babe590e4ead3870236\nChange-Id: Ieeb5aac0c70dd1b87b27379cd61cc52d9767c508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/934618462cc7a571a5c1c17ea502155b2227195d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/934618462cc7a571a5c1c17ea502155b2227195d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
