)]}'
{"id":"openstack%2Fpuppet-nova~598819","triplet_id":"openstack%2Fpuppet-nova~stable%2Fpike~Ida7a0e4241617034ecd5f816b87beac85aed482c","project":"openstack/puppet-nova","branch":"stable/pike","topic":"python3-first","hashtags":[],"change_id":"Ida7a0e4241617034ecd5f816b87beac85aed482c","subject":"import zuul job settings from project-config","status":"MERGED","created":"2018-08-31 12:57:00.000000000","updated":"2018-08-31 19:42:30.000000000","submitted":"2018-08-31 19:42:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"598819-1535744550160-221bfd5c","meta_rev_id":"101bf970fd86629c547348e67160df55a5d6c796","_number":598819,"virtual_id_number":598819,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2018-08-31 19:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2018-08-31 15:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2018-08-31 15:47:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-31 15:47:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2018-08-31 19:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"963ed43cbb21ce7c9e31007ce620ac9c03af7a35","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-08-31 12:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66ad031c0875287c8b27d5a8966dc6b83f63a230","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-08-31 15:47:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f5dbab0ad350be0c0ac51c5a9920e46603b823c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 17:46:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/19/598819/1/check/puppet-openstack-lint/aab4fab/ : SUCCESS in 4m 24s\n- puppet-openstack-syntax-4 http://logs.openstack.org/19/598819/1/check/puppet-openstack-syntax-4/e8a1e42/ : SUCCESS in 3m 56s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/598819/1/check/puppet-openstack-unit-4.8-centos-7/0209e47/ : SUCCESS in 40m 35s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/ca85a68/ : SUCCESS in 1h 09m 56s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/1c757e6/ : SUCCESS in 1h 01m 49s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/ffbe5ae/ : SUCCESS in 1h 11m 07s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/047737e/ : SUCCESS in 44m 55s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/8fecc22/ : SUCCESS in 31m 24s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/eec6541/ : SUCCESS in 37m 36s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/5f15a74/ : SUCCESS in 32m 00s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/7d4d996/ : SUCCESS in 30m 48s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/19/598819/1/check/puppet-openstack-beaker-centos-7/bde3ef9/ : SUCCESS in 22m 53s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/19/598819/1/check/puppet-openstack-beaker-xenial/066f6d7/ : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"543adb260f98f5aa566e59dd6d848410936008e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 17:59:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaffb51bfbba05fd16777eac60c87fc25f7ac2b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 19:42:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/19/598819/1/gate/puppet-openstack-lint/6537b73/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-4 http://logs.openstack.org/19/598819/1/gate/puppet-openstack-syntax-4/25d1de9/ : SUCCESS in 3m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/598819/1/gate/puppet-openstack-unit-4.8-centos-7/05d996e/ : SUCCESS in 42m 36s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/ae55843/ : SUCCESS in 1h 04m 46s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/b112a74/ : SUCCESS in 1h 07m 57s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/82f29d0/ : SUCCESS in 1h 02m 45s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/12c6213/ : SUCCESS in 46m 22s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/4a1ec8f/ : SUCCESS in 36m 32s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/f087e7c/ : SUCCESS in 41m 10s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/c08c424/ : SUCCESS in 31m 26s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/19/598819/1/gate/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/c60c424/ : SUCCESS in 32m 17s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/19/598819/1/gate/puppet-openstack-beaker-centos-7/e35c793/ : SUCCESS in 22m 19s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/19/598819/1/gate/puppet-openstack-beaker-xenial/2142454/ : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"acb3395676dcf72497393a5eedd8b77da3c9411b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 19:42:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c847cd3ddb92a7b3668e45034ac7286be5f2fe6","revisions":{"7c847cd3ddb92a7b3668e45034ac7286be5f2fe6":{"kind":"REWORK","_number":1,"created":"2018-08-31 12:57:00.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/19/598819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/19/598819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/19/598819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/19/598819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/19/598819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/19/598819/1"}}},"commit":{"parents":[{"commit":"a8b7530e0297d1bc98cfd1d1989efcad8b6499b1","subject":"Allow live_migration_inbound_addr to be used with non-default port/user/extra_params","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a8b7530e0297d1bc98cfd1d1989efcad8b6499b1"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-31 12:33:47.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-31 12:33:47.000000000","tz":-240},"subject":"import zuul job settings from project-config","message":"import zuul job settings from project-config\n\nThis is a mechanically generated patch to complete step 1 of moving\nthe zuul job settings out of project-config and into each project\nrepository.\n\nBecause there will be a separate patch on each branch, the branch\nspecifiers for branch-specific jobs have been removed.\n\nBecause this patch is generated by a script, there may be some\ncosmetic changes to the layout of the YAML file(s) as the contents are\nnormalized.\n\nSee the python3-first goal document for details:\nhttps://governance.openstack.org/tc/goals/stein/python3-first.html\n\nChange-Id: Ida7a0e4241617034ecd5f816b87beac85aed482c\nStory: #2002586\nTask: #24325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/7c847cd3ddb92a7b3668e45034ac7286be5f2fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/7c847cd3ddb92a7b3668e45034ac7286be5f2fe6"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"a8b7530e0297d1bc98cfd1d1989efcad8b6499b1","is_merged_in_target_branch":true,"change_id":"Ifdc5fbd05195604ab6ea6564d0905f9385c6df67","change_number":567831,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
