)]}'
{"id":"openstack%2Fpuppet-nova~772175","triplet_id":"openstack%2Fpuppet-nova~master~I6c627855d6c90ae8ad18898fb8c2adb8d73477e1","project":"openstack/puppet-nova","branch":"master","topic":"cleanup-neutron-opts","hashtags":[],"change_id":"I6c627855d6c90ae8ad18898fb8c2adb8d73477e1","subject":"Clean up the deprecated neutron_* parameters","status":"MERGED","created":"2021-01-23 23:42:16.000000000","updated":"2021-03-01 11:42:42.000000000","submitted":"2021-03-01 11:40:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772175-cleanup-neutron-opts","meta_rev_id":"d15bd52050a7051222b07abf4faac1e0b0e8abda","_number":772175,"virtual_id_number":772175,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 11:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-02-19 14:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-02-23 17:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2021-03-01 08:31:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-24 02:36:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 14:38:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-02-23 17:55:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"6de6799ebaa20b172470e4491bb57e5e03b23722","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 23:42:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6adbec7c284ec4cedda8c300cb389f85df1cbdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 02:36:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55d1f048be46419590b2855cc2490d8d : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fc7d83fbdbd4531a6016a98e2f8d15a : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/eb9214a3dfde45cfa75f399772131b8b : SUCCESS in 13m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e2156a4ebe74dd181a603802fa232dc : SUCCESS in 13m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/979073bf280647928a16c06c866f5ea1 : SUCCESS in 14m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9b50d4bd13d4301aea8453065bd2384 : SUCCESS in 39m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f695e3da1ad4e8b860a94cb185bccc4 : SUCCESS in 52m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abba24cc86a64a788fdf5d8702c39204 : FAILURE in 45m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd54f02a619e47d7a83460e6734e26d4 : SUCCESS in 32m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/787515f54e574e81acfa31abee251c5e : FAILURE in 10m 00s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6c9be10266144b5bb0972848ce6b5c97 : FAILURE in 8m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/04306645fca1442ab1edca77f88d2694 : FAILURE in 6m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d61fb6e79c9343c89b2a6737aa567ac4 : FAILURE in 10m 01s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/74d7f2d327124548a133e0e14204655c : SUCCESS in 30m 21s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fd37cf392de4617a15416a5722fbdd7 : SUCCESS in 20m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60c141a914c34e638f9d777d15277d04 : SUCCESS in 4m 04s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f3b0915904e645858f77ea5a218695b8 : SUCCESS in 2h 52m 44s","accounts_in_message":[],"_revision_number":1},{"id":"78622672a563e09633e5067937c1fe6d0e5e8bfc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-25 01:12:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8628214944a8e47e8aaf4cdca158f7aeb88fbccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 04:33:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6361127cfd734221a58163a358b2f503 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afec84c97b384ecdb6e4d7f28667e05a : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3aef1f5123d448fb89bdef6e035fe96d : SUCCESS in 14m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f012e0f4b6554f848d8586dbaed53b9f : SUCCESS in 12m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/062a2c4152a84c4b91ffa75e4d61fc12 : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bdb927b4ead4b5eb407dcde9f07b3b0 : SUCCESS in 56m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aedb65a4be574b6f8bf06ef7d6e7b6f2 : SUCCESS in 56m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cd469424ac042f19496c259efe24393 : FAILURE in 42m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f04663624d24244b450a8930fb0c507 : SUCCESS in 42m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d9a575037a7a4cf99c99670b053aeac8 : SUCCESS in 1h 05m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9d2c6286bd14f168449268c9782a99b : SUCCESS in 1h 47m 55s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aa8309b387024a43ab3e5c940ff9e6a8 : SUCCESS in 1h 29m 33s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2487c036b4eb499da23b71fec51cfeb9 : SUCCESS in 1h 12m 25s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/40c16ed1609948b281b3810cc505e377 : SUCCESS in 31m 54s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/191ddb714a0d47b7b29a1d5f41fe2935 : SUCCESS in 18m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46cff7e9cd6a4633ae1a409d25d39512 : SUCCESS in 3m 19s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b02b928ce0dc485b85e6014043ec5745 : TIMED_OUT in 3h 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"61557330b3c0024656626281277c0e792fd35dd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-19 13:50:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3dc85392c5803a29b6581e3410d8c2e9f69827be","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-19 14:38:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5a15378a13e70b12228f97c300ca0966ad891636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:56:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c737eddf8a1440359204d3a409d1d7f1 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b98f62cdfac4ce4bc65ac5669639f2e : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/075ad2083ed34127aede0b7e658b4981 : SUCCESS in 16m 12s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9f9c355bae64549ab2d8904e67e1d9c : SUCCESS in 17m 11s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ac239f903fc4f93927a577d01c1f22a : SUCCESS in 14m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1d68a27c35049c181eafe5fbe9076ed : SUCCESS in 17m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/876ca724af23497d94e62c8274adac56 : FAILURE in 36m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a983b2f45b1434cafeba6cb3f071d27 : FAILURE in 34m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/195b347c64a7492580c502991b8572b0 : FAILURE in 30m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d03ed620b3443adb90f04cb5c28681b : FAILURE in 41m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/62085b7e2d5645568acf634da80ff37f : TIMED_OUT in 2h 00m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4867efac25714920b38242ba416dd801 : SUCCESS in 1h 45m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/667cd3d113874bf584cf5926d9975e33 : SUCCESS in 1h 27m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cbbb9076e01241deb235c842013a3d72 : SUCCESS in 1h 04m 02s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43f6917fe5d94b84bfd21dcc839ea571 : SUCCESS in 1h 15m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc0538acff7e4d089ce42a64f99d6f2c : SUCCESS in 1h 23m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e5184b95b9f43139b715f1ae6b357b3 : SUCCESS in 1h 22m 28s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc38f2bdfcd94306a72bf1eb82bc7a89 : SUCCESS in 1h 09m 32s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7e68ef83f773434a80f9de58b046634e : SUCCESS in 24m 25s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd801621e0f344018f9b0c1d6be91c99 : SUCCESS in 19m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/726b6a37472f494d946997fd230a94d6 : SUCCESS in 4m 42s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9961c2ab269c4f90afbcbc5eb70ba966 : SUCCESS in 2h 45m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3d181682194e057e9e45157005f2d08caa3c1bdf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 17:55:10.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fc028265def55ccb88b87f414b9562eb3470a6ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 21:31:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8c5cd1e5efe4baa87915b80cb14e901 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d64f41d5f824342a9efe78864d79b4b : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cd4fa1c584334f49bd279f7a999c7290 : SUCCESS in 15m 26s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/442dc68988f448b3ae3ebbee27825906 : SUCCESS in 16m 49s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4862dbdefeff4431bab209dadf89714b : SUCCESS in 13m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e88ad992d51240b78a98560bd9fb088c : SUCCESS in 17m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4180acb1aad4225917c7b38c7934e6f : FAILURE in 35m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49fbd8c7607c45839ed1b1b91aec6ef4 : FAILURE in 34m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6a3a930a4164ae4a6aee959024f665f : FAILURE in 34m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13fefad0d36941eb8ec3411e496f66cf : FAILURE in 32m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b74616afb784482bb90fa1a2a947c51 : SUCCESS in 1h 20m 05s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8af7bbf810b447729b4923c9e2369b1e : SUCCESS in 1h 29m 23s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/75b9bd57940a40008beb4eb3ccc843e9 : SUCCESS in 1h 28m 42s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8ef5ef26e34441ef9b40d3db69bfd72c : SUCCESS in 1h 04m 57s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a03397527df49c08ee828a5c8185476 : SUCCESS in 1h 16m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e7a764ffd084424aef9fec83046bfcb : SUCCESS in 1h 26m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/02f195a4a7544ffb97b06dcc05eb8723 : SUCCESS in 1h 22m 00s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5b855a3de694d9a969336753aa64fbb : SUCCESS in 48m 12s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c87c8f0907634caaa27f9db80770fc4f : SUCCESS in 23m 07s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62f167a711a34fa99366e347d6b9c07a : SUCCESS in 21m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/287706b958ab4b08870d8c73fa4d75fb : SUCCESS in 6m 27s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/33a0a471c6a34e5f8a030c67a0844a3c : TIMED_OUT in 3h 14m 35s","accounts_in_message":[],"_revision_number":2},{"id":"545433cbeb4486a77cfd04776aa9dda271ac10ed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-28 06:47:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"10ecb8cd2792de5921ebce408a9c341e6cdeb6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 09:59:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23674a470f2a4cc3b5326960f29aa0fe : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f97c856b324499a956002045270667a : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/43c217d718574f4187788158ca9fd880 : SUCCESS in 14m 33s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a61480929374b66b15a245f0a01488c : SUCCESS in 14m 52s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c8bc532a3ff4ca7bc8f49aeced7ec5b : SUCCESS in 12m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c268c7e56d9c4a12b1998139edad9d32 : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef764b57556b4c7a8b09b2502c7df9fb : FAILURE in 48m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb5dbdc0c11443f8ab30962cbd6c6906 : FAILURE in 47m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0a67fa2d5f7497a9bbf2debaa985ce5 : FAILURE in 43m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6192c3869ca477d91cb2c1553b7103f : FAILURE in 27m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4e8bb7daa7624e05af93f78a4ded761a : SUCCESS in 1h 30m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/001f49a6ccb941d1b21a7dd815ab807f : SUCCESS in 1h 43m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a68db395e8d8433da702b20c42b4e0bd : SUCCESS in 1h 46m 02s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c6e8e9ebb221496390b4d44d60ca7487 : SUCCESS in 1h 05m 15s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/255a6ad705414c29a29173c90b4c27bd : SUCCESS in 1h 09m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6137819ae76f4d58bcb53ebb04a6eb8a : SUCCESS in 1h 31m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/85c30e2c052b46f5bc46b0f34e8ed3f7 : SUCCESS in 1h 27m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc21fe9a419044068c4ab16dbdd567d6 : SUCCESS in 1h 02m 50s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/bc52396a77b243be89c672eaa3f25775 : SUCCESS in 29m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1575d1d289049408983d25bbaef8e74 : SUCCESS in 21m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/437c4c4fbe8940edab5c9f9bc26e633a : SUCCESS in 4m 18s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3eecd79aadc244018fbe0b753225b97a : SUCCESS in 3h 04m 43s","accounts_in_message":[],"_revision_number":2},{"id":"bd0d46a85e96063c0f8ad2f4618d5132c7d4e691","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-01 08:31:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4ab4de7d778b2b65bd3a41262968eef17f16ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 08:33:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c73d569970efbdf8c4c53124e08a3046797303d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:40:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca8f16cdabc34bf0922122d80234a766 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4af6565a27c4411a9c74741ba904efeb : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d90cadcc613d43fdb79b0ec075f2b804 : SUCCESS in 16m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90a5677e0f1c4e3b928aa4386ccfe708 : SUCCESS in 14m 40s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9fc9403cecd64550a4e105b0b187d1fc : SUCCESS in 1h 33m 51s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ec868134dd2642919f11af2f1fea9e2f : SUCCESS in 1h 59m 08s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e7f7468145648bbb6cde9904193c997 : SUCCESS in 1h 32m 13s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/734289c89933482da9258734b0a0e52c : SUCCESS in 1h 15m 32s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0bef2f3681c74f46ab5eb70278d0d22d : SUCCESS in 25m 54s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24dcc70034a94165a88ccaba777b64ea : SUCCESS in 22m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19bed696712540e58e36aaf554df9ab2 : SUCCESS in 4m 21s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/389604a296d242349da4df87dafcb8c1 : SUCCESS in 3h 03m 53s","accounts_in_message":[],"_revision_number":2},{"id":"7da53278a910e7d2f7e06b39235c1c5321b36188","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:40:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d15bd52050a7051222b07abf4faac1e0b0e8abda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:42:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/232faeec716041bf8b325e744165f331 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d18ad5bcc7358db9c71f1326f8b6a83784fdf725","revisions":{"8f9f5785dd7b1f02644d309ed88dc65cfb0978a2":{"kind":"REWORK","_number":1,"created":"2021-01-23 23:42:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/772175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/75/772175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/1"}}},"commit":{"parents":[{"commit":"614b847449936d16a5b753bc3cfef3c3afea8234","subject":"Add parameter to set image_type_exclude_list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/614b847449936d16a5b753bc3cfef3c3afea8234"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 23:41:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 23:42:07.000000000","tz":540},"subject":"Clean up the deprecated neutron_* parameters","message":"Clean up the deprecated neutron_* parameters\n\nThis change removes the deprecated neutorn_* parameters from\nnova::network::neutron, which were deprecated during the previous\nVictoria cycle.\n\nChange-Id: I6c627855d6c90ae8ad18898fb8c2adb8d73477e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8f9f5785dd7b1f02644d309ed88dc65cfb0978a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8f9f5785dd7b1f02644d309ed88dc65cfb0978a2"}]},"branch":"refs/heads/master"},"d18ad5bcc7358db9c71f1326f8b6a83784fdf725":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-19 13:50:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/772175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/75/772175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/75/772175/2"}}},"commit":{"parents":[{"commit":"8ab07ef0a5ce6c62827b0d9cfcbeedefc9144d8a","subject":"Merge \"Do not re-define defaults for quota parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/8ab07ef0a5ce6c62827b0d9cfcbeedefc9144d8a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 23:41:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-19 13:50:23.000000000","tz":0},"subject":"Clean up the deprecated neutron_* parameters","message":"Clean up the deprecated neutron_* parameters\n\nThis change removes the deprecated neutorn_* parameters from\nnova::network::neutron, which were deprecated during the previous\nVictoria cycle.\n\nChange-Id: I6c627855d6c90ae8ad18898fb8c2adb8d73477e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d18ad5bcc7358db9c71f1326f8b6a83784fdf725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d18ad5bcc7358db9c71f1326f8b6a83784fdf725"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
