)]}'
{"id":"openstack%2Fpuppet-nova~800961","triplet_id":"openstack%2Fpuppet-nova~stable%2Fvictoria~I227f40111591cfe1026ce992b3de3e28d092f329","project":"openstack/puppet-nova","branch":"stable/victoria","hashtags":[],"change_id":"I227f40111591cfe1026ce992b3de3e28d092f329","subject":"Add parameters for image copy when using multiple RBD Glance stores","status":"MERGED","created":"2021-07-15 15:24:55.000000000","updated":"2021-07-19 15:51:47.000000000","submitted":"2021-07-19 15:50:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800961","meta_rev_id":"97446a87237c8f7c6fa51b210aa3bc4911448b3a","_number":800961,"virtual_id_number":800961,"owner":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-19 15:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-07-18 09:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-07-19 13:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-07-19 13:25:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 15:32:45.000000000","updated_by":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-15 15:32:45.000000000","updated_by":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-07-15 17:37:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c98ab29243cf931aa7c180fc03dda2302ffefe41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"date":"2021-07-15 15:24:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"680feab6e4eb1056fa5c358a30339f7bfb12b1bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:37:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b48241ab511f4323a02164401d095d5b : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c923f18d1c64826a3bd2e9171def05b : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ddc609682fb4524a63fb900984b26e7 : SUCCESS in 17m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b4e5db44ce8406b99576f58f04230be : SUCCESS in 14m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ffdaaf54eb14c14bad3edcbc14cb237 : SUCCESS in 19m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe86c73d07b2412bbbc3032491723ac3 : SUCCESS in 49m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9bf317724c4422cb483982dd8568acd : SUCCESS in 1h 01m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95dfb26724e7419689079dcd76b523e8 : SUCCESS in 47m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/517aaad2f2974a9fab6a524db353f3d7 : SUCCESS in 38m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3c62a4ac5ddf4123bb7af85f3efd206a : SUCCESS in 1h 13m 04s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/37cd1b1938bf4e869b378ef4b0612d95 : SUCCESS in 1h 17m 56s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b50fc8f486904d6e99c09aeca4d3cf2d : SUCCESS in 1h 39m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a06a011b6f5a4d0abb91a0762985bcd6 : SUCCESS in 50m 23s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b7a39bf6e7242f3bbdb2d06b7904457 : SUCCESS in 1h 09m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/53f6c11552af48799dc61d799dee7995 : SUCCESS in 1h 58m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e121dbab6a84605af6ae7ef07694065 : SUCCESS in 1h 27m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/600c892dd5eb4114a38af4e4a2dea9ee : SUCCESS in 53m 29s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c46b3514bb26474a9339815d0fe0c3d8 : SUCCESS in 28m 29s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/00353181c50d43f484a6111531dc1b10 : SUCCESS in 27m 55s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06d05416505c453a9b401c3b4b574de6 : SUCCESS in 21m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84b7309ba60346cca95cc9239290bd95 : SUCCESS in 4m 10s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eb97a61274a648dc8cdbf89de9f6501b : SUCCESS in 2h 07m 41s","accounts_in_message":[],"_revision_number":1},{"id":"241b4bbe368c2b220c70c3b4400a8381923ee3a7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-18 09:59:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26fa244acaee67035a1188c006eb0c8389e857c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-19 13:25:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f48733d2d70bbaf901baa2c23eae285db8782eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:26:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c292a05482d9b0055832cbe999b0d94061dd4588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 15:50:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e7132358c9a4b58b419b18ddbe12731 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e0edb9eb0ac4ce69aca44011c5c0276 : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a0f55c66a5f40a7bab222b0d57b14b6 : SUCCESS in 16m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/078132d9b0244a93a896211409be23a3 : SUCCESS in 15m 40s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e8c10aeac9194ae2aaa17f279c9b090e : SUCCESS in 1h 39m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/57ff597d5fac4e67bf41575ba7c85c68 : SUCCESS in 1h 24m 22s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/87c35003ecc640b59887915ebdfcc75f : SUCCESS in 1h 45m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bd21f137dcf24dbabbab3aa036f14fde : SUCCESS in 57m 22s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9bb58d65b5694599a890c0f333ad9993 : SUCCESS in 27m 14s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae270142f0874d71a6c1633372c371ae : SUCCESS in 32m 18s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36fc55d1d53a414687d91552402f0cbf : SUCCESS in 21m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b3a4acc67be4e379885a18e85367adf : SUCCESS in 4m 30s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/89dfb3ab80694114a005b4a35d3206d9 : SUCCESS in 2h 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2a62073c4fb5c21b8b9a94b52085f89bb14b4940","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 15:50:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"97446a87237c8f7c6fa51b210aa3bc4911448b3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 15:51:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8edc53a16f2e470b9c3df6a8b60a33b3 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fab23231c466c18b0cc872ee46a458644c81601","revisions":{"3fab23231c466c18b0cc872ee46a458644c81601":{"kind":"REWORK","_number":1,"created":"2021-07-15 15:24:55.000000000","uploader":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"ref":"refs/changes/61/800961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/61/800961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/61/800961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/61/800961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/61/800961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/61/800961/1"}}},"commit":{"parents":[{"commit":"0700c1159cf5dd2f134fdf36c5b2206b472f6205","subject":"Merge \"Victoria and before: Use libvirt-daemon-system in Ubuntu Focal\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/0700c1159cf5dd2f134fdf36c5b2206b472f6205"}]}],"author":{"name":"Oliver Walsh","email":"owalsh@redhat.com","date":"2021-04-15 12:55:40.000000000","tz":60},"committer":{"name":"Oliver Walsh","email":"owalsh@redhat.com","date":"2021-07-15 15:24:03.000000000","tz":60},"subject":"Add parameters for image copy when using multiple RBD Glance stores","message":"Add parameters for image copy when using multiple RBD Glance stores\n\nExpose the Nova config options to allow an image to be automatically\ncopied into the local Glance store when Glance is configured to use\nmultiple RBD stores.\n\nRelated Nova change: Ia839ad418b0f2887cb8e8f5ee3e660a0751db9ce\n\nChange-Id: I227f40111591cfe1026ce992b3de3e28d092f329\n(cherry picked from commit 920f2fbf7e1d455b0f8ab0e7655dc576e0642cd7)\n(cherry picked from commit 885da96c6685b721402a3cd2228559cb90ec81c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/3fab23231c466c18b0cc872ee46a458644c81601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/3fab23231c466c18b0cc872ee46a458644c81601"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
