)]}'
{"id":"openstack%2Fpuppet-nova~871404","triplet_id":"openstack%2Fpuppet-nova~master~I9d533275d5999fa02b438be1642f61b40a246161","project":"openstack/puppet-nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-25 01:10:48.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I9d533275d5999fa02b438be1642f61b40a246161","subject":"Fix inconsistent parameter name of wsgi::apache","status":"MERGED","created":"2023-01-23 01:32:59.000000000","updated":"2023-01-25 05:10:02.000000000","submitted":"2023-01-25 05:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871404","meta_rev_id":"7b37a040a15ad13aa480134a84d87ac00a43d681","_number":871404,"virtual_id_number":871404,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-25 05:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-01-23 17:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-01-23 19:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-01-23 19:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 03:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 17:58:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-01-23 19:35:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0ccccd8bded00fa3de4362fe3e205d92a58ba0ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 01:32:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1e92393c9a6a2fca08d120a78631530607fb814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 03:46:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbeef0b1cfa64d58bdaee3ee9d8d407f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3300e560d6146f5bd5dbe4151606e47 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57a6c05a67584e27b04575888fcab6fd : SUCCESS in 3m 06s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bc66839fe64418e8ca32d6307cd1dce : SUCCESS in 16m 24s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05313a91681d464ea607054538096e0d : SUCCESS in 13m 16s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b3b09ad689e4f52bca6620498c663ee : SUCCESS in 12m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cddb90a0e8c84df086a63b6f981b1c2a : SUCCESS in 51m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a60042e0ea624718a3496697b9677d73 : FAILURE in 54m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33e7cb30459549ebabc25f8b69ea0191 : FAILURE in 32m 21s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b65a4885c5843c892a2f8d1fd42995e : SUCCESS in 33m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe1ab6cc95074204ae946d8a21c3fd46 : FAILURE in 40m 13s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b54fe1fba15241f1ad41062b23eb35d5 : SUCCESS in 23m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/65056f6e1bf2456ea59aba01749e863e : SUCCESS in 48m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a21cf5c628f4fb5b99e51a0c88583b2 : SUCCESS in 1h 15m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84c080f44b264582b334c9192de7ce8e : SUCCESS in 1h 10m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b11feeba0084ee0b315b199292a1ff8 : SUCCESS in 54m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b56703c0c7d74ef18aad035df242077e : SUCCESS in 1h 00m 42s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a3b11165cb24bc88c82375f0edcb70a : SUCCESS in 25m 29s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f9f3c814b74bc98a17b7aa0b169d2f : FAILURE in 6m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c11a2c8195a46be8bddbcdd2e9ecef2 : FAILURE in 7m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13fca65fcfe447d18a16b9b78f7a5127 : SUCCESS in 4m 40s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/17f6fbf62db0430ca91dff93ead602ea : SUCCESS in 2h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3ca7d83481fa17bb3115b37fcecc6b67da27b67a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-23 17:58:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e866f617bff7dce5feb53d8419a7f4ad3112baee","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-23 19:35:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92c5bdf3d04013756531ca3715c72fb4641f1b9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:36:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b48aa8e200351bad8f72f2163faef18412cc3659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 21:52:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e507a98bb6df4a26a0b225132e96e938\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3edd552086a47a6a33f9a918e1be000 : FAILURE in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/464f59eab5f14401ba8695be0f7b7a87 : FAILURE in 2m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d7becfb143249bb931d4a0fa91876dc : FAILURE in 8m 07s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb3123d83fe943f6b82eae9648563da8 : FAILURE in 5m 01s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/815b1690625b4f18a7fe1618e8c6d76e : SUCCESS in 51m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ea21490c4c7494d8b0e2df4133594ec : SUCCESS in 1h 10m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/55da3da26a2047a7a8f10c6677c21440 : SUCCESS in 1h 03m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b28dcd756ea40659c230fb437e0039d : SUCCESS in 1h 12m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99d78eaeb9054dc0baca96935e5a8001 : SUCCESS in 52m 14s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5718bc7c755446c6af9f7e5cffd22f67 : FAILURE in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80a4e4c83ac94faba87da9747231ce7a : SUCCESS in 3m 59s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ea2ae386565f4566b3851fbe51836c48 : SUCCESS in 1h 47m 03s","accounts_in_message":[],"_revision_number":1},{"id":"885e0eda39d9d969f10adb4761d15d49f94f2f17","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-24 01:11:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9da16bf2074718bfcf7ff67fe0906552113c9391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 03:23:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f8961e8314d4f89ae22df659ff9a236\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf4ba3e9a62c48da8cb0e65505579889 : FAILURE in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/750d69f253204159b73bce5e2f4c46e4 : FAILURE in 5m 24s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3882c3032e7b4d0dafa3d1f9a8c9fa26 : FAILURE in 7m 28s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec3819af7209414f99e1999390a5f111 : FAILURE in 5m 10s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9ea51fa5b2e4c8e82823356cbadd839 : FAILURE in 4m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/437a977dbe6a43b08abc8b6d7048eb8b : SUCCESS in 52m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63b0f41226d8416c82f6284344935183 : FAILURE in 53m 22s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56434cd8934842a69136156d4ce6030e : FAILURE in 24m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b011607b9f074b8a81f383ea5939b571 : SUCCESS in 41m 40s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1e6bd1601f34461806254ea4417f251 : FAILURE in 34m 36s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d60b16ad66924e219ebe7621c23205d9 : SUCCESS in 16m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8be53386a1e4e91a39b9f4ada175204 : SUCCESS in 1h 02m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fe4eb2b90bb46c7a0fa0ce3c39cbeaa : SUCCESS in 1h 07m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25a242a939e2466e99ff5093698718de : SUCCESS in 48m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8d72bd405914cb5ad1fbbab23cb0e69 : SUCCESS in 55m 17s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/658c5523dfb249cb812a005c0df01052 : SUCCESS in 51m 07s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7113a4fb674242ad9fe88f4cd1db3921 : FAILURE in 7m 34s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59d7e04e5c294830ad7a6f8cea8448f0 : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23b201caad414f38bd98314ff2d22a00 : FAILURE in 5m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/703a4c0c0dc8429f9e5ac4b6aa2b5c36 : SUCCESS in 3m 48s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/23563413f29641cc8387f4430939b7d7 : SUCCESS in 1h 56m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e713bec383ada71fb358d304fff44330922a2b32","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-24 13:52:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b12f5a4a27192daab751d8e606bcbeaf34a209da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac5832f7041740dfb54dc915e464f2ec\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7edc3017bf9c42f089af444baad0b344 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae1dcec2a15947ddae06555c59ce6afe : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d25ada8886442c386cc5e55a43b6b48 : SUCCESS in 15m 41s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d654b110e92648c7ac4408af8a8d4b88 : SUCCESS in 15m 16s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afdea98b7d744b01b81c82b590f73921 : SUCCESS in 14m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/097bb4a6804548988fef4207a80128ab : SUCCESS in 48m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc657e70370c4df0a09139f7d546cacd : FAILURE in 57m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2230e1f60ae40d4967aaf1f826eb020 : FAILURE in 27m 21s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c059cf4cd40b4105a0f4bd6ffd7991d1 : SUCCESS in 41m 07s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82df44d748ba40688b6f96f35cc3b4fc : FAILURE in 39m 52s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e09e0a6df1b843b3984d69622143f81c : SUCCESS in 16m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b8c795c73de49cd82e825b2ecd06b3e : SUCCESS in 1h 01m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9f60884a0d54a40b4d550cb2fc5b32d : SUCCESS in 1h 14m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/87ff3e9f30d3401780d56b8a9e6a870b : SUCCESS in 58m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bf7d8f39fb141a7bc5092cb556047ee : SUCCESS in 52m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a17269ea4f8c4459b469fa30f67b6f5e : SUCCESS in 54m 24s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fee785a0c8641e7a9cbcf2347833c2d : SUCCESS in 30m 13s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5dc095eb49448fbbce80f00c161a7cb : SUCCESS in 21m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dbd0279c0474414afc739f49965df70 : SUCCESS in 20m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/980c0fdf24ee46cf950b7b0e0e3cbe6d : SUCCESS in 3m 09s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7c66db16a60a4befa89c6ab9a5ab15e0 : SUCCESS in 1h 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c6f5b51f6996901dbd1a0bdcc4396c35e36672dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:02:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae79eaec5af029ba7a9de638d789245ba4e21888","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:50:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80e69f692f5140988755fee267cc01f1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24760da490d64501aa9238cfc6b069a7 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d03f42acc6a942cda50c759cc7a747bd : SUCCESS in 3m 08s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/132df1b8e1f042cf86060659eac4c9ee : SUCCESS in 20m 47s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06321810ad264463b8160f9318fc4214 : SUCCESS in 15m 12s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fab613555147420b95765c8d5f930971 : SUCCESS in 58m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d23b360a16bb415ab16aa23862a41401 : SUCCESS in 1h 15m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66d2003988ac4c0082117a7e5e010eb5 : SUCCESS in 58m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7253fe777cde4242b655b54a6d70ed37 : FAILURE in 52m 45s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2287c2639bf4352b97cd3b1e1da1b7f : SUCCESS in 55m 26s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ee6eb3960364f24966acaf087bfbf82 : SUCCESS in 25m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93556febf16743b6b24602f8168d461a : SUCCESS in 3m 35s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/83aa474ffa854a43a5ac3580d2084159 : SUCCESS in 1h 37m 29s","accounts_in_message":[],"_revision_number":1},{"id":"69db999bd8148fd572ca9f18bb3b06d66c55c43c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-25 01:10:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ed3ebf0f125946284f07fb3cc624c89308440f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 03:04:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16090ef578f0420ebd492288893734ea\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2584580d30ce4d99b31583ad3679e544 : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcd4b7d097884fc3a7ffae9cc7d3ca25 : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d2d7db9b5b349e084596dcc14f3a3ea : SUCCESS in 24m 25s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6ff206fb30a4ceeb7cb1647e9c50573 : SUCCESS in 16m 04s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e9c1eb72f394be4b16796cb2589be10 : SUCCESS in 15m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c94e9ca28d89455982c645f30721b3f4 : SUCCESS in 55m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47789d14bb6943688cd0168864e96ee6 : FAILURE in 1h 06m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/142a095cd1964a65ae3a3ca32621bf46 : FAILURE in 29m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a928e0cb8cc415695cd48ae37882cb7 : SUCCESS in 45m 54s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9480b48edd074283949ef3eb5b6d289f : FAILURE in 35m 53s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c57fb8de01e64d61869cce532af5ae20 : SUCCESS in 17m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b40ca963f4749e2a2c19e7d6dde667b : SUCCESS in 52m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c55cdcd9a9ab40d0b067822b7e8bcd2f : SUCCESS in 57m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6cf9e2cd3384965a796fad175182454 : SUCCESS in 1h 10m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e51a022987e24c04b7597659e0f6da47 : SUCCESS in 50m 03s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d90e4ce3fae41f1b84254f49b26b545 : SUCCESS in 54m 08s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f749edd9886245d9906b9a79de407c2b : SUCCESS in 27m 54s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/305c406f65a343b69ef495191f855146 : SUCCESS in 23m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f4846ffdbec474091b784846791cf13 : SUCCESS in 19m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/304558885feb4c198859d4c843825e41 : SUCCESS in 3m 14s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bdde8eba802a4bf088e2beb51d381b47 : SUCCESS in 1h 47m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8a5365d5db0e56cfbbff10860faea1e7fe39626e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 03:05:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb27c83bd103f04c68bf179273c0a916955169f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 05:08:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d66f9017f2b41c9b3d63f42b29b23d8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5bc240cc8774eb189701586bef4ab87 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b2af1f246834766b69147786358d1db : SUCCESS in 3m 35s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/04289725547749b4aca901ca3029bab3 : SUCCESS in 19m 52s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d140ec256674870abe12f22180943c8 : SUCCESS in 15m 05s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dbf3d76475a47da87e27c6446bb9c44 : SUCCESS in 55m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad8e120249504c6c95481a2c177162d1 : SUCCESS in 1h 07m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/824a191b3dbd4063802a992a9af3a5e0 : SUCCESS in 1h 00m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3ff816b83e6499d8d6a51865c25a8e0 : SUCCESS in 52m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0b1eae78bb54e23a2f7fdbbbf535e25 : SUCCESS in 53m 11s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6025a515abd649d79fb38beedfe92e44 : SUCCESS in 28m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/804c5e5f9dad420ca93aa499d0075f7d : SUCCESS in 3m 54s\n- puppet-nova-tripleo-standalone https://zuul.opendev.org/t/openstack/build/78af8a64ee5942d3a5a417e21d27dcc9 : SUCCESS in 1h 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b1cde8c1f80d85ca3de13de9160cfcc54db1a87e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 05:08:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7b37a040a15ad13aa480134a84d87ac00a43d681","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 05:10:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a72d756a28e46b69293318a3d2c50f5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d07f09e4017040f0b821985f2c973759 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a258d3d8c76859d7084129b3e369ac9004fc5c82","revisions":{"a258d3d8c76859d7084129b3e369ac9004fc5c82":{"kind":"REWORK","_number":1,"created":"2023-01-23 01:32:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/871404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/04/871404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/04/871404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/04/871404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/04/871404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/04/871404/1"}}},"commit":{"parents":[{"commit":"2e0276d34878293a140815fad6376f6c3ef6f55e","subject":"Merge \"qemu: Ensure boolean values for tls parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/2e0276d34878293a140815fad6376f6c3ef6f55e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 01:32:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 01:32:48.000000000","tz":540},"subject":"Fix inconsistent parameter name of wsgi::apache","message":"Fix inconsistent parameter name of wsgi::apache\n\nThe nova::wsgi::apache class takes the api_port parameter to determine\nthe TCP port number, but this name is not commonly used while the port\nparameter is used in the other modules. This change fixes the parameter\nname so that we can keep the implementation more consistent.\n\nChange-Id: I9d533275d5999fa02b438be1642f61b40a246161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a258d3d8c76859d7084129b3e369ac9004fc5c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/a258d3d8c76859d7084129b3e369ac9004fc5c82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
