)]}'
{"id":"openstack%2Fpuppet-nova~888275","triplet_id":"openstack%2Fpuppet-nova~master~Idda7645ebb0bd5b59ec44028585aca600ff9868e","project":"openstack/puppet-nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-07-18 11:28:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Idda7645ebb0bd5b59ec44028585aca600ff9868e","subject":"Deprecate ineffective periodic_interval","status":"MERGED","created":"2023-07-12 13:23:22.000000000","updated":"2023-07-18 14:20:57.000000000","submitted":"2023-07-18 14:19:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888275","meta_rev_id":"9a80c37e47ab0a25dbccccd013c7bda8693529dd","_number":888275,"virtual_id_number":888275,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-18 14:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-07-17 16:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-07-17 23:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-07-17 23:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-12 15:02:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-17 16:23:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-07-17 23:37:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b30e0fb07c582b27570fa4d43fadfeb50a2397c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 13:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5413252f3fe9f76c6e901a04c7aacec065815d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 15:02:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfcfba8516594f4da032043f7a0d1168\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb1245c5613642bda4280f2ce17e3eb5 : FAILURE in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9b39cae1b5e4469b741cd4664ab7e15 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/95b955ca25484c9e99d10ae0b3021821 : SUCCESS in 5m 23s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8b92bbaa721402eaa5683a4e2775de4 : SUCCESS in 16m 20s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1b07dbcf2f1466e937d18a1117e1651 : SUCCESS in 14m 33s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/08a1dd02c97845dc93f65e2cc2d91374 : SUCCESS in 18m 35s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/786e20826a084092a5697deafbed1ae4 : SUCCESS in 15m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c78406beed4648b3a376d3b6b6a8fe76 : SUCCESS in 41m 38s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54dfb1dd40244cc994f415189f7ddfb9 : SUCCESS in 56m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a1381586c454a679a20c5fbb6f39fd8 : SUCCESS in 47m 55s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43d745f122a847ac8ebdf16d3e5cdc04 : SUCCESS in 59m 51s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dc7c15cbab94a8383d8a6480b594f74 : SUCCESS in 40m 53s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df035d4b62334484ace9c8a71ac06a5d : SUCCESS in 22m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fadd69f3a55041dda57ad7f82d9f499c : SUCCESS in 1h 07m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a48b6129886141ad90176ed6988de38c : SUCCESS in 1h 17m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f607b3c04bf24f6d9f29c390ad6dcc48 : SUCCESS in 1h 04m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c02635d1d4c54725842f57d14c27d9ad : FAILURE in 50m 18s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f19726f9543447a4a614720cb76a0dd3 : SUCCESS in 57m 46s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/09dfc20519424c37af572af70b5017f7 : SUCCESS in 25m 08s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2475baf707441259aa8f88b4cf647b6 : SUCCESS in 21m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faadd961f12d4ddc8ee0acb55b4a9678 : SUCCESS in 22m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0c464e0b2e0408cb88894952daa0800 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"96f333ceba84ced2d21c29a4069d377663fe90f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 15:37:08.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b113eba61b3829b49f98f4e27119ee5a314a9a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 15:59:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"84ad8ddbf36aa535588fcb3a0a0c988f16e16ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 17:34:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/497df98474a540f7b8334aa54ce181eb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a667bda3873147179ec7f322fb22496c : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cab796c3663045d58afe6280e71d9c51 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a833b0201da4ce4bd0ede88b58521d8 : SUCCESS in 6m 01s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d811341141a471ab0c0279dbf43fd89 : SUCCESS in 19m 08s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3248ee58069c46559b5c178033a50b0c : SUCCESS in 12m 25s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4bad170457043818f8921f54970681e : SUCCESS in 17m 21s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d070697fe959486981407951be0618fd : SUCCESS in 12m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc7e79810a4b4d959d1547aa6d4700c6 : SUCCESS in 49m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a24afe9a24d942cba50d7fc98c4e0611 : SUCCESS in 1h 00m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc81fa7ecf344bf88ca5813a42e9ec79 : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/377e93274c0e4238ae079a7499d49090 : SUCCESS in 40m 28s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/485fb923acde42bf9d648b1a0ebd5891 : SUCCESS in 44m 04s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/047d8cb152b549e990461bdc49bcec9a : SUCCESS in 21m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd7d4a6413c448ffa98776ddb7136f91 : SUCCESS in 1h 08m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49865ed236d24b8eade2e25da3e2fc11 : SUCCESS in 1h 25m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c277696401645719ea3f4a431e46367 : FAILURE in 1h 21m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eacd3289cf7343bb884ad680c3a1c69c : SUCCESS in 1h 01m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/77760826cafd411da220d74fc0462f0e : SUCCESS in 1h 02m 00s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/54fca9cfbd5f41229464b0cd14441ef4 : SUCCESS in 31m 59s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e02bb6f90dd04e5180f430438f1ef087 : SUCCESS in 22m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed88ca7dd10744b7a110462aa14220fb : SUCCESS in 22m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba439b58f7dd49af9be2c7269a8aca13 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"383d88e16d09841496267e41cfbb280c32cb7e08","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-17 03:58:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a55feaf0277290abe59c2f7273d57524d899a3ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 05:41:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b69ffbf1a54ab890714e9bdb6d374b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1181bcc2f61d4566b67311439b6d6998 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16e6164da65549ed888f9e8af8306cb5 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dfe3806132f4e9a986ca186190063a9 : SUCCESS in 5m 06s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/42fd8f6e57a345aeb831c1f3f78a70a2 : SUCCESS in 14m 19s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dd8d3b080c346579868e03d7a57fcbf : SUCCESS in 13m 47s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcb323c447014a8382831758c7ff06ba : SUCCESS in 20m 06s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97dcccde981d45b78c30119b0a27099e : SUCCESS in 15m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ad42f1eae9d48f7af99dd7205855978 : SUCCESS in 48m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64608ba3a2004e5ab1d37877f51346fa : SUCCESS in 1h 05m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a6feced996b4f0aa8cb365c241e843d : SUCCESS in 43m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20dad844e03349b79b6f4166825f1075 : SUCCESS in 54m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4dd9de29d0d4614b30f9529220fdb55 : SUCCESS in 43m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/132896be1977426bbcf50202a475ad64 : SUCCESS in 25m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0ab58d1c47c49a1a99e4bab32605c21 : SUCCESS in 44m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4307784a9094979a8f0f8a1fbd1b189 : SUCCESS in 55m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43bee12088734938a00d221b381714d9 : SUCCESS in 1h 11m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/899ddd8c723547deaf7b800c7ade9f19 : SUCCESS in 46m 49s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9a1a7c5a7df4a0ab20257604818521c : SUCCESS in 46m 28s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/89462a4c8f5c4ebca9adf15b9d6362bd : SUCCESS in 28m 03s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68e6c91c07d6432d9ef3e118418374e1 : SUCCESS in 21m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e0f2afcd9b840a78bde7d55a4fd9bab : SUCCESS in 22m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37897a2a7b0b47829adc579801c51d4f : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"7d5b27b7a663de6dea313dd7492a91739b12cc13","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-07-17 16:23:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"35b5ca3f4865f630df42c1efe900a8c50dcc5685","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-07-17 23:37:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a7e88fc7729bbf872c0710ec5f004c34974d59fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 23:38:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4733bbd1c148fd15d67f335854d6f1354f8ed638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 01:04:52.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/862ffce3086c457686766b200cf6fd1d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43df7c3a430748dbb371de98abc184ae : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20998208de514f10838518435f1623bc : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce68e489fa984942964fae0685933c3f : SUCCESS in 14m 06s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d41ef1b39d7462f83256e7997266354 : SUCCESS in 10m 30s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d1019a646d446b298a2dd55a7e45d39 : SUCCESS in 17m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f58a3ea7db14919a929520dc8b73543 : SUCCESS in 47m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff223f7c8add49dab1648cd09f62238a : SUCCESS in 1h 21m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed4c79d3662f4eac84eec72ae5e30776 : SUCCESS in 51m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca8bf5be67024bbfac7c52861848e6c5 : FAILURE in 48m 30s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04ebe18f891b4e13bb24d42eabaaa934 : SUCCESS in 53m 42s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdf2ba9ed6e84b2a8dc2dabc57d4dc8a : SUCCESS in 22m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bb6e1abf30d453a9cd59671a92203b7 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"19d9fef5516005744f5d89937ecd3867f96b359a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-18 07:08:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ec595d58e6a757981f39152b2cc231be9d1ae41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 08:16:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d80303da3e24f0ebee20ac04aac29b8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d1f3ab27e444b9296b2b882e2fa0c7a : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97ee354720b542c182282440f9668091 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/81849c79e369499cbcadd78bfa9cb9aa : SUCCESS in 4m 17s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1570432c9d194544a37aef0513c9f6f5 : SUCCESS in 13m 01s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10b0cd16706a4a7fabf32e02354d9f07 : SUCCESS in 9m 37s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/361907a980894c2b8d83ba49c80f21fe : SUCCESS in 15m 22s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5047b37b254848559bef0773065a7489 : SUCCESS in 14m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc1b2c48d23148c6bd7f9498a38ebc19 : SUCCESS in 48m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e703ed0edcab49598d9db2887e7560a1 : SUCCESS in 51m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18955a9fd79444798db5df138af4c4f0 : SUCCESS in 37m 25s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de56d95b6efb4fb7815218f18b3b87a3 : SUCCESS in 39m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22d62d561c3c48058cbe94ade574e4b7 : SUCCESS in 36m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/960e969e3d0a448b86fe179618686239 : SUCCESS in 18m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b25dac94768465195a7eb25bc2eb2c8 : SUCCESS in 59m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f6222b38aba4e6b83a5de5dadde2f3c : SUCCESS in 1h 02m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8502eb5430354a31856019d79b7194b0 : SUCCESS in 1h 01m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9cb7fb4017f40ac83b5a756c48186de : FAILURE in 56m 11s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2dbe2b510be4c809e548e4da54741b9 : SUCCESS in 49m 39s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c207b9273e94d85b31593f356b8a395 : SUCCESS in 25m 23s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0d75fec95e249dab237e86fc3d15eba : SUCCESS in 16m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6fb6aeb31f74b1e8074f5af76ac9318 : SUCCESS in 20m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ec1f53014b240a6b9c161cd27b01b00 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f34f5ea0ce646fc8ecff04e9f27a3607f50afe45","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-18 11:28:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07451ee8517c2ce89c2fbb55300c0e8a8585c39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 13:00:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46cae0a4b99c48d88cab7844d5f4380e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de32d9084066485fbeff71250e347b10 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b68c3a3c8b248588cd800504d80fd3b : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7eb7bd8bcbd449f49b4d86ef0b9dde83 : SUCCESS in 4m 30s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7643944c635f48f38beaff3cb0598475 : SUCCESS in 18m 54s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be5c7c59e6284c28953f99e5e112d27c : SUCCESS in 13m 27s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/104d2b33ccea47009bef0856f6a65dd9 : SUCCESS in 13m 56s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af211ef7030547a59fb96f1152e71545 : SUCCESS in 14m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc70263d136c43b3b0cb52529717b9c5 : SUCCESS in 58m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c378636e2fe84836b54f4e132cd4a463 : SUCCESS in 53m 24s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bc3a291f30145d4b7eec26532dea60c : SUCCESS in 56m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7295344503184b5db79c6261763487fe : SUCCESS in 38m 27s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3625bf3c7f454265a901b2c5aae86a08 : SUCCESS in 44m 19s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/499c236d33584aa5bc31a50d81a05050 : SUCCESS in 25m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/90175172a7a04fab99dc7e65ffb6ed0e : SUCCESS in 1h 10m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fba7bea6eeca42bcabc17cb786984d97 : SUCCESS in 1h 09m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/16dd7d751da4491eaea266b12f92d70c : SUCCESS in 1h 24m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/092e09bca8ff481b9bf9f5356ae8b7bd : SUCCESS in 1h 03m 04s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/832721b3ca2d40a4bab98c8de3946fb1 : SUCCESS in 54m 53s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f91703c56fe4205bc7421371f48c216 : SUCCESS in 33m 16s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de93e9f486c844809fe49c0aaaf1caae : SUCCESS in 22m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76d866db4bae4a7187bd3c4b48e9c480 : SUCCESS in 19m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/422315b9db054583a7cb8f4e0720d0b0 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a20fd0da0f6fce15f33bb6a208f8ea4433635a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 13:01:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e30beb587bcca76a5048a88dc174c739b4db83a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 14:19:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9facbd0582742f2b29b3aa174c52a7c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7e0558f4cff41ab9806f9c707990c80 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b220bec2475c41fb9124446a2d999274 : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/487e032a8967447db49609337a95a5cc : SUCCESS in 15m 54s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9ca18e52dac4df0911606e925195fe1 : SUCCESS in 16m 00s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/985360ad95d64271ba6f2d1bb1128467 : SUCCESS in 20m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cce6426f098495aa191e5d13069ddcd : SUCCESS in 1h 09m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/affe29d549c6405699a2ea22585f7b6d : SUCCESS in 1h 13m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6436717d23ae4efd8b1ed3562b2ed5c2 : SUCCESS in 1h 02m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a5d7baab2124aa0b41c72380851c414 : SUCCESS in 52m 49s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/83ceb77cc5da489ea7d424e9b47c6f6c : SUCCESS in 53m 42s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1ca3064415f458bbd29bb3fb9c70364 : SUCCESS in 23m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5c1a94d38d043a9a79e2e1c7f54ed6e : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3},{"id":"2c24c02ad947f685c3d6de8dfb3750dc4edf8c68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 14:19:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9a80c37e47ab0a25dbccccd013c7bda8693529dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 14:20:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdedf962cd3b456ea177731a3188c49a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e648c522828244fab71420845c276715 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9d97a006344aff606b8265a5ebf34f66ea1ff152","revisions":{"84f9e11a8f177bd6424e802671f658a3578fc9cf":{"kind":"REWORK","_number":1,"created":"2023-07-12 13:23:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/888275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/75/888275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/1"}}},"commit":{"parents":[{"commit":"d844dcbd50f98a2426d50f0a263756f10d91f922","subject":"upgrade_levels: Deprecate support for cells v1 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d844dcbd50f98a2426d50f0a263756f10d91f922"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 13:22:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 13:22:13.000000000","tz":540},"subject":"Deprecate ineffective periodic_interval","message":"Deprecate ineffective periodic_interval\n\nThis parameter was not implemented properly. Furthermore nova does not\nsupport the periodic_interval parameter.\n\nChange-Id: Idda7645ebb0bd5b59ec44028585aca600ff9868e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/84f9e11a8f177bd6424e802671f658a3578fc9cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/84f9e11a8f177bd6424e802671f658a3578fc9cf"}]},"branch":"refs/heads/master"},"2bd14f7469567cde5712c3c539a6c87e541e88e7":{"kind":"REWORK","_number":2,"created":"2023-07-12 15:37:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/888275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/75/888275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/2"}}},"commit":{"parents":[{"commit":"d844dcbd50f98a2426d50f0a263756f10d91f922","subject":"upgrade_levels: Deprecate support for cells v1 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d844dcbd50f98a2426d50f0a263756f10d91f922"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 13:22:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 15:37:08.000000000","tz":0},"subject":"Deprecate ineffective periodic_interval","message":"Deprecate ineffective periodic_interval\n\nThis parameter was not implemented properly. Furthermore nova does not\nsupport the periodic_interval parameter.\n\nChange-Id: Idda7645ebb0bd5b59ec44028585aca600ff9868e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/2bd14f7469567cde5712c3c539a6c87e541e88e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/2bd14f7469567cde5712c3c539a6c87e541e88e7"}]},"branch":"refs/heads/master"},"9d97a006344aff606b8265a5ebf34f66ea1ff152":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-12 15:59:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/888275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/75/888275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/75/888275/3"}}},"commit":{"parents":[{"commit":"adbb5ce857554050cf48dd57c7aeade04248a4d1","subject":"upgrade_levels: Deprecate support for cells v1 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/adbb5ce857554050cf48dd57c7aeade04248a4d1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 13:22:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 15:59:50.000000000","tz":0},"subject":"Deprecate ineffective periodic_interval","message":"Deprecate ineffective periodic_interval\n\nThis parameter was not implemented properly. Furthermore nova does not\nsupport the periodic_interval parameter.\n\nChange-Id: Idda7645ebb0bd5b59ec44028585aca600ff9868e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/9d97a006344aff606b8265a5ebf34f66ea1ff152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/9d97a006344aff606b8265a5ebf34f66ea1ff152"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
