)]}'
{"id":"openstack%2Fpuppet-nova~901683","triplet_id":"openstack%2Fpuppet-nova~master~I99ddef51ef60ae1b1bc6c43643cce7596457b598","project":"openstack/puppet-nova","branch":"master","topic":"libvirt-guests","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-27 17:09:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I99ddef51ef60ae1b1bc6c43643cce7596457b598","subject":"libvirt-guests: Improve parameter coverage","status":"MERGED","created":"2023-11-22 17:09:02.000000000","updated":"2023-11-27 17:09:20.000000000","submitted":"2023-11-27 17:09:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901683-libvirt-guests","meta_rev_id":"63979fd1e0817a7aa7d3b7a2bcc52413864ac1bd","_number":901683,"virtual_id_number":901683,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-27 17:09:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-27 09:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-27 15:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-27 12:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-27 15:28:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-22 18:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 09:11:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-27 12:28:27.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-11-27 15:28:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"200c36d84cc70c646dbd7c4e816a0ec4e789696a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-22 17:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b1f461a960b76bcdfc006c99618019d1934b7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-22 17:10:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54520743c19d582c35fda9cf4db782308ce0eb4c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-22 17:13:27.000000000","message":"Topic set to libvirt-guests","accounts_in_message":[],"_revision_number":2},{"id":"d1e8823888ca2ee4a86c2fbd5b0a9ecfa87b60d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-22 17:24:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aeeab9a8eb1c82fb6727d956acf80684e627ca0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 18:53:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69a3f5893cfe4e548039ef617cb00638\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cc9a08b4e6d4a9cb1e32f5e26d28681 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3731eb93abd04694a5b9405ab0b65ee0 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a96a41705124a58942f267b81d32c8e : SUCCESS in 6m 04s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa2a1b09eaa745a2bc5de712b5af4a20 : FAILURE in 18m 49s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1688050b90a49839f89fb00b6f31f13 : FAILURE in 15m 48s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6eb845397cc414f8187e4fa2e6b5a26 : FAILURE in 19m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a727bff74df49b98169d62515b55e36 : FAILURE in 12m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c2dea68dcef4b5bb803cababb4d2328 : SUCCESS in 49m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0af62b841b04bbcac28a8797790bc17 : SUCCESS in 56m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40ab68ce37094e0496aa46229efbf920 : SUCCESS in 59m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50d91be2a5694d5da6e514bf9d2a4098 : SUCCESS in 43m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/137fef20569f46aab6b71b2e778b8b6a : SUCCESS in 43m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddaef25346bd451292a54e7099527cbf : SUCCESS in 53m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6451d9217a2343eeb89e511d8914e417 : SUCCESS in 1h 28m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fda32d98068743448286e81e96208043 : SUCCESS in 55m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a152505665648868a8d7b5e8516b8be : SUCCESS in 1h 08m 09s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d8dcb32fb3945229f9b7c0d33f00fcb : SUCCESS in 48m 14s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/761a797b7e2a4f8e988fcc57a67bf638 : SUCCESS in 27m 55s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/892daf1d67614c0faf32ffce728da10b : SUCCESS in 18m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/174228cc6e83490e800f4bd0af7896d2 : SUCCESS in 19m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a381362d554bd89913acbd3eef0d1b : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c28163fec25700957a5da6f5a4a2d330b3828c77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-22 19:16:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a792a73ef94db77698caee4f07bbfe6ae6253062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-22 19:20:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75b0682d3b202cbe7307d3e10aff5d97d21ccdfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 20:51:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf1ee838ea0b421f84e88c94b7ca26fb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/370ec02049034ddba10037397ca96b11 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1953500378b04fe38fa8c22797f3f58a : FAILURE in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3ba5e8d48cf477aaaa31f0a17666fd0 : FAILURE in 4m 49s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b62414d10d94c7d9fb180ce5d127779 : FAILURE in 17m 35s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7b50210c7094a9897dc55c434a042d4 : FAILURE in 14m 28s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/edba1087e26c4d1c9a9a9ec02f373aee : FAILURE in 14m 38s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4483c4206633429a9bea08aea6e47c36 : FAILURE in 14m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f78bcad299714075927c9f8d3919c625 : SUCCESS in 46m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca46e7b12f3f4b69bcdf585af16dc409 : FAILURE in 52m 27s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7887ce4cad4546d592414e8890eb2ef9 : SUCCESS in 41m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46af4127f52740e3ab36a03a489925e6 : SUCCESS in 36m 19s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/261b6888497b42ac8f49340ad0c66e76 : SUCCESS in 42m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9cf74fb0d8d45d7a67e80b2b1e8bf4d : SUCCESS in 1h 03m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1018f3db9842461dba121c5113e8b216 : SUCCESS in 1h 22m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc4e4238b91d4d1aa6ef30bb05d116a9 : SUCCESS in 1h 03m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/645c738723e9421ca8e30b4dbb912ca3 : SUCCESS in 46m 56s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8d336c1c75346c59db21407e3cf3258 : SUCCESS in 45m 48s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/53a2006ab3954677b68e54773e07752f : SUCCESS in 21m 28s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86200eb6951043019c7b0eecb228816a : SUCCESS in 18m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8cd881394974a269e963c066faeb7bd : SUCCESS in 19m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afe8d48e62094aa086e3d6af36d59105 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":5},{"id":"056bd2f127d29704db473d047780aecd08e5b5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-23 07:02:43.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"faf6c6723454c205bd49679ea43555c837703d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-23 08:29:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a4e142833157523f834dd609272e5f6904e84d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 09:59:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0826e14d4cdd461a97cd88d11cd1809e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8625671ef274ccdb2a598b84df0cea8 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e51f29e50a7746749bc2ccebb7836f1d : SUCCESS in 3m 05s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a78e397ff6a4150b9816fb7d13ea3fd : SUCCESS in 4m 27s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/e93191bd209f499ea4f1c58419df2c80 : FAILURE in 13m 15s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1f4a3f047234e86b4d7e2e7173d24dd : FAILURE in 11m 27s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5b57ea930f3474e873cb55a2406cda0 : FAILURE in 15m 17s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bfdcbe3e6a04dd299dcb9418f59aa6b : FAILURE in 14m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65cb720e9ed8428187ff937f60764067 : SUCCESS in 46m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8b9b0e8eb2b4778baac634f087ece24 : SUCCESS in 57m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4361900ac2754e0ea286e6462c4f2e03 : SUCCESS in 54m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e537c9917294b53936f8a7786903904 : SUCCESS in 45m 33s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29879f28aef64c2bae7b795f70b3217d : SUCCESS in 46m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffacc52ea2ba4f649f7d71af5aa9ae6e : SUCCESS in 1h 00m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0a393a6610e4f438ebe787115406001 : SUCCESS in 1h 23m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/631031492ce0477998f551a6297e5bc2 : SUCCESS in 1h 04m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d87f9de60df0410dac8054bea4352522 : SUCCESS in 55m 50s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/34ce62ce8e5848b38af6530d4fbdb21a : SUCCESS in 44m 32s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/d109c115e5b444e0b36bb79976289686 : SUCCESS in 27m 18s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c520c6e22534c3bb0f06f0ee6f77749 : SUCCESS in 22m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7f6cdfbeb424adfa1610d31916fb4c9 : SUCCESS in 19m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/714c892f85084042b7b66b3046e26e78 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":7},{"id":"5e23bff5a3f846fb33ab680bc963ee37dbb10ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-23 10:57:09.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4e922b37da43f56a8f3e6b5b6be043949684ba9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 12:12:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/477cb4701cb24123978081371f8c5cdb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fde7cd6be2974ef59d4a8381bea9221c : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d006f3afbe34466ca7af1966851ab90b : SUCCESS in 3m 51s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ae657fd4d384ed2adbc37ffd0f4716d : SUCCESS in 6m 07s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/29d5d1ef9df74e19a4e5b7a23f19c325 : SUCCESS in 18m 06s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07df4e94841b4da7845e8603f1c7c92c : SUCCESS in 14m 10s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/4364d3d271fb4e0faba7057cac4c4db3 : FAILURE in 18m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0719bca46191456c93f5c3ee18d7ab78 : SUCCESS in 14m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5ff1e24f926499a9dcb3d2643684fe9 : SUCCESS in 47m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d717fdf49d984fa2bb44dbe268fc1f2d : SUCCESS in 48m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88893efa709941eaa6dfccb869b9a389 : SUCCESS in 52m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12331b3eb1e446d9bcc92842e17e028d : SUCCESS in 44m 13s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c77d583f21a4236ade6301ebd587f6c : SUCCESS in 38m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/62dad0406e084b27874f1bd564e4c45c : SUCCESS in 51m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0af33e82de7943668e58231de6c88a68 : SUCCESS in 1h 08m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04688fbd5bf7492aab820b69fd121418 : SUCCESS in 1h 06m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfece5bd0c0f41ad87a17c36106c2d3d : SUCCESS in 59m 07s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8adb57a3cd5147d6a7c07ec9a139ecfb : SUCCESS in 43m 33s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec68e8f29b84457bba6983ef1f7eb4b6 : SUCCESS in 22m 24s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1afcb080c8c4688a9ed3e5fd4b17245 : SUCCESS in 16m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ee0b9b571fd46b69a237ad94d77861f : SUCCESS in 17m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b95969ca1414788a674599d24859231 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":8},{"id":"e9f3b1577975899e5323284d94e4f44f9351ebcc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-27 09:11:02.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"78e498c87b93154ee308ec22f355c07eb5d201b9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-27 12:28:27.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"977c2bec0e939f8c40ab2cfbb0547a49bfc13ff1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-27 15:28:16.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"251624b84840b85bb855253396c1fc1cefe43a08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 15:29:09.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"32329d0ac40235cd729dc531fb97a90f4ef6599b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:09:20.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a219c7bfc0dc4c61a57203a9faa81265\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/379765d8e9bd474ebeaa531c3ab92cb9 : SUCCESS in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/388727c88ce743dcaf5e8baf419c10a7 : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f85d6a684bd4d97978a26524e098aee : SUCCESS in 13m 21s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f94b15733c1745abb271a57ae1c07374 : SUCCESS in 15m 27s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce93c89cfd8346529dfb550011cba00c : FAILURE in 18m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8922e24110c14c2d8025a6a92d926a26 : SUCCESS in 1h 03m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c3870a2729142a6bb25af021b4952dc : SUCCESS in 1h 08m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be08a4e5d0e143baa03a3837ad0d27fe : SUCCESS in 57m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8839a04e669442ca474454025f30b94 : SUCCESS in 1h 01m 13s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c77795fa54554a8bbbe688889e5a1156 : SUCCESS in 53m 16s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/28bf59c640da492d95a9d1755e64d19c : SUCCESS in 23m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/636117c130ac46a78bf064361831daad : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":8},{"id":"63979fd1e0817a7aa7d3b7a2bcc52413864ac1bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:09:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d88dc6659c45f1011a09ecb05794eb35f7ac834d","revisions":{"94dc22ddc00e6cb5008344f2ef3e9455fb6056f6":{"kind":"REWORK","_number":1,"created":"2023-11-22 17:09:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/1"}}},"commit":{"parents":[{"commit":"07d0681eef497d1101987431c481202561301da8","subject":"Refactor libvirt-guests support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/07d0681eef497d1101987431c481202561301da8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/94dc22ddc00e6cb5008344f2ef3e9455fb6056f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/94dc22ddc00e6cb5008344f2ef3e9455fb6056f6"}]},"branch":"refs/heads/master"},"738835a05925cb4fd36769bd8a6ea5973ca54391":{"kind":"REWORK","_number":2,"created":"2023-11-22 17:10:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/2"}}},"commit":{"parents":[{"commit":"07d0681eef497d1101987431c481202561301da8","subject":"Refactor libvirt-guests support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/07d0681eef497d1101987431c481202561301da8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:10:29.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/738835a05925cb4fd36769bd8a6ea5973ca54391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/738835a05925cb4fd36769bd8a6ea5973ca54391"}]},"branch":"refs/heads/master"},"f159a68176d4d81cf884a2dc8a8385fd3a4d3d74":{"kind":"REWORK","_number":3,"created":"2023-11-22 17:24:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/3"}}},"commit":{"parents":[{"commit":"07d0681eef497d1101987431c481202561301da8","subject":"Refactor libvirt-guests support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/07d0681eef497d1101987431c481202561301da8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:23:38.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f159a68176d4d81cf884a2dc8a8385fd3a4d3d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/f159a68176d4d81cf884a2dc8a8385fd3a4d3d74"}]},"branch":"refs/heads/master"},"fca79522c6536d7dd535c72c22f54155df8011cb":{"kind":"REWORK","_number":4,"created":"2023-11-22 19:16:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/4"}}},"commit":{"parents":[{"commit":"eea360fb9715f62e19b55d3d7f90507e758ab9b1","subject":"libvirt-guests: Remove hard-coded default of shutdown_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/eea360fb9715f62e19b55d3d7f90507e758ab9b1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 19:16:04.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/fca79522c6536d7dd535c72c22f54155df8011cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/fca79522c6536d7dd535c72c22f54155df8011cb"}]},"branch":"refs/heads/master"},"e61b6c6e29c361a4dcb612b1eaea56946bab5319":{"kind":"REWORK","_number":5,"created":"2023-11-22 19:20:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/5"}}},"commit":{"parents":[{"commit":"0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b","subject":"libvirt-guests: Remove hard-coded default of shutdown_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 19:20:25.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/e61b6c6e29c361a4dcb612b1eaea56946bab5319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/e61b6c6e29c361a4dcb612b1eaea56946bab5319"}]},"branch":"refs/heads/master"},"ceb1d42b3716543a9bc6d207e2ab52551a5e5a9c":{"kind":"REWORK","_number":6,"created":"2023-11-23 07:02:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/6"}}},"commit":{"parents":[{"commit":"0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b","subject":"libvirt-guests: Remove hard-coded default of shutdown_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-23 07:02:31.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/ceb1d42b3716543a9bc6d207e2ab52551a5e5a9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/ceb1d42b3716543a9bc6d207e2ab52551a5e5a9c"}]},"branch":"refs/heads/master"},"6e2d725c31f9cd68771bac1389d587b0b05c543e":{"kind":"REWORK","_number":7,"created":"2023-11-23 08:29:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/7"}}},"commit":{"parents":[{"commit":"0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b","subject":"libvirt-guests: Remove hard-coded default of shutdown_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-23 08:29:45.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/6e2d725c31f9cd68771bac1389d587b0b05c543e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/6e2d725c31f9cd68771bac1389d587b0b05c543e"}]},"branch":"refs/heads/master"},"d88dc6659c45f1011a09ecb05794eb35f7ac834d":{"kind":"REWORK","_number":8,"created":"2023-11-23 10:57:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/901683/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/83/901683/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/83/901683/8"}}},"commit":{"parents":[{"commit":"0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b","subject":"libvirt-guests: Remove hard-coded default of shutdown_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/0c2b228bf8be5a6b51ddf8a29e67ca65c67e532b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 17:08:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-23 10:57:00.000000000","tz":540},"subject":"libvirt-guests: Improve parameter coverage","message":"libvirt-guests: Improve parameter coverage\n\nThis introduces support for a few more parameters in libvirt-guests\nservice.\n\nChange-Id: I99ddef51ef60ae1b1bc6c43643cce7596457b598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d88dc6659c45f1011a09ecb05794eb35f7ac834d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/d88dc6659c45f1011a09ecb05794eb35f7ac834d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
