)]}'
{"id":"openstack%2Fpuppet-nova~938380","triplet_id":"openstack%2Fpuppet-nova~master~I4d217ccc843ab62f8fc7e8c56755e7c93f1c884a","project":"openstack/puppet-nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-05 22:13:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d217ccc843ab62f8fc7e8c56755e7c93f1c884a","subject":"Drop unused constants for tgt package","status":"MERGED","created":"2025-01-03 02:15:55.000000000","updated":"2025-01-05 22:13:54.000000000","submitted":"2025-01-05 22:13:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938380","meta_rev_id":"b5408369d7fab4fc103fba50d8c361cea275741c","_number":938380,"virtual_id_number":938380,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-05 22:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-01-03 21:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-01-05 20:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-01-05 20:53:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-03 03:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-03 21:40:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-01-05 20:53:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e7e31c65aedccfdb5215166f91f1fee60b4bdc66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-03 02:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64f449e4f3f0cb8ca0b8a973d713297dfce82fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 03:54:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d54d076e9b9547e7b5bb3c43ef5756ea\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb8c72c01bb64b34a2c8505b5e1eccb3 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/415750d7fddc486b8b711c3749354bd2 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e798a726f6c430abadc319b65e88cf4 : SUCCESS in 6m 18s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fa08051b2a24b1aba53e1e69d131ba9 : SUCCESS in 15m 39s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/755bd985dc1b472bbeb5be06585175e6 : SUCCESS in 16m 32s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/880e201cdd2a4ead9d83007ce715a031 : SUCCESS in 16m 52s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebe14e25a80f4406b87a8993cd0fd65f : SUCCESS in 14m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9a48158288249edbacddfaaff86096c : SUCCESS in 50m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/403cf6cb63c345dd90c178e322c950ab : SUCCESS in 1h 03m 02s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb3c60793c56490282e3219ca0910dab : SUCCESS in 53m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a703ce745c894a8e9c111436751e5b6f : SUCCESS in 43m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51d4a782e2b54f8f9d7bbdd27d621fb2 : FAILURE in 30m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e356673a34642d3937267c1e835a00b : SUCCESS in 1h 03m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a44c6406e1ad47c498e3c8fdd92fc55f : SUCCESS in 1h 19m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/827db3259d824e2d8a119c792c00b890 : SUCCESS in 1h 01m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbfd86afdce94fb29d0e1ffb705d340f : SUCCESS in 53m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cadaedebadb4d04802168480957e42a : SUCCESS in 1h 13m 43s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3518c3ba398a478bac9fddd337138c52 : SUCCESS in 27m 23s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38a75b967681487a85270e9a0ff62eed : SUCCESS in 17m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34d10bd1945545e89ce813b08bfa6ab5 : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa6352eb6eef8e25b1cedf6b478d69b1eced9fa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-03 05:36:19.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d36f98634e2c4d4462cb33e377f41c906097992b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 07:06:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b05d226bda4337a232cf95e45e68f6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3e66881003443f8984c331fa677dd02 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d07f9951379a404fa96983273faf4fe2 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/689ced455f44440ab00c3a9465b63660 : SUCCESS in 4m 19s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0aae93be6d7b4d43b89bdc10083b929b : SUCCESS in 11m 26s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aabe75baa104c439ab2457c92374e5e : SUCCESS in 8m 25s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b5061026b0742d489682a256a46977d : SUCCESS in 11m 34s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27e504d206f646b38729bd80e7abf096 : SUCCESS in 8m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/945b50e40c2543e48b66032081ad825e : SUCCESS in 26m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9f3b19d9ee04e76bc929964dd7638ac : SUCCESS in 30m 38s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/844af34247144973b53ccdba7c84df8d : FAILURE in 30m 24s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c64b7dc3c9cc4fe9b0f5a5069356bcd9 : SUCCESS in 23m 50s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/591d0c7fa5754e17b43fc115f8fab61f : SUCCESS in 37m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9192f5d9e4840cd80cf2cc7474dcba7 : SUCCESS in 36m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f00a95a20ce846d6acb0c6cb53c4ac1f : SUCCESS in 1h 24m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3efd3e085a5c471ab2acb6722939a4a3 : SUCCESS in 1h 03m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc7f58bc2bee4d6ea8968c386e2b4545 : SUCCESS in 1h 00m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a663123825648d4844226f2eec434b7 : SUCCESS in 1h 01m 46s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3815cf2355444983a1a33f3be44c8013 : SUCCESS in 26m 46s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26d50357ca464c0c9fdb12f813b625dd : SUCCESS in 18m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10df9bc6cd3d4e16af54e92ea1443744 : SUCCESS in 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f969f341d21b4980c5c4fe8ee9b0be550069171c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-01-03 21:40:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1e579a90fd2f3bcfcba8ae90a221483fb5e59650","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-01-05 20:53:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"370a4027db699697e7ccd0b1dc15ae0f5cb765be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 20:54:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ceb4f6ec7bc1e6a604ce40a1ef6708c143c48b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:13:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aabb1200de546a5875aeadb577e52f0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b24034f62b046ac9031c5cb02db0000 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dea38de392a341ff85bd9769f914d9fc : SUCCESS in 4m 23s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/33640c83b40543d1973ddfad73ac5e71 : SUCCESS in 17m 23s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0486a1f6b57348c5b1c46bb430e7d9db : SUCCESS in 14m 03s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9043d982447407b993ef7f302b184c6 : SUCCESS in 1h 02m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7a3fe0bb8484615b199df31597f59ed : SUCCESS in 1h 00m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5491d3d1361348baa499cf77fbe70a3d : SUCCESS in 42m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4487fa8780e94664b68caf27b70053ed : SUCCESS in 48m 13s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17658a45ab0942269819d4b4017665e2 : SUCCESS in 51m 40s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c71dd7570b845c98d7c2240092bf031 : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b5408369d7fab4fc103fba50d8c361cea275741c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:13:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"caa18f690260717222d535687d465efc225b3a1e","revisions":{"233d0981b87fd99c75843633e42c22a97af3040f":{"kind":"REWORK","_number":1,"created":"2025-01-03 02:15:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/938380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/80/938380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/1"}}},"commit":{"parents":[{"commit":"4044c4d5985e8970dc9215f02472ac382c2e62b6","subject":"Merge \"limit: Add default service_name/type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/4044c4d5985e8970dc9215f02472ac382c2e62b6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-03 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-03 02:14:49.000000000","tz":540},"subject":"Drop constants for tgt package","message":"Drop constants for tgt package\n\nNova does not directly require tgt package since nova-volume was\nremoved (it was replaced by cinder).\n\nChange-Id: I4d217ccc843ab62f8fc7e8c56755e7c93f1c884a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/233d0981b87fd99c75843633e42c22a97af3040f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/233d0981b87fd99c75843633e42c22a97af3040f"}]},"branch":"refs/heads/master"},"caa18f690260717222d535687d465efc225b3a1e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-03 05:36:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/938380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-nova","ref":"refs/changes/80/938380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-nova refs/changes/80/938380/2"}}},"commit":{"parents":[{"commit":"4044c4d5985e8970dc9215f02472ac382c2e62b6","subject":"Merge \"limit: Add default service_name/type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/4044c4d5985e8970dc9215f02472ac382c2e62b6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-03 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-03 05:36:19.000000000","tz":0},"subject":"Drop unused constants for tgt package","message":"Drop unused constants for tgt package\n\nNova does not directly require tgt package since nova-volume was\nremoved (it was replaced by cinder).\n\nChange-Id: I4d217ccc843ab62f8fc7e8c56755e7c93f1c884a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/caa18f690260717222d535687d465efc225b3a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-nova/commit/caa18f690260717222d535687d465efc225b3a1e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
