)]}'
{"id":"openstack%2Fpuppet-octavia~597725","triplet_id":"openstack%2Fpuppet-octavia~master~Icd17bfb5addf37d11207d008bd59583f929cb6d5","project":"openstack/puppet-octavia","branch":"master","topic":"puppet-octavia-testing","hashtags":[],"change_id":"Icd17bfb5addf37d11207d008bd59583f929cb6d5","subject":"Add scenario001 to zuul config","status":"ABANDONED","created":"2018-08-29 22:44:27.000000000","updated":"2021-10-11 09:01:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd6ff808577025515fcb69416b1a70efa9720143","_number":597725,"virtual_id_number":597725,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":-1,"date":"2019-04-25 00:22:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-06 21:19:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-01-02 10:07:50.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-25 00:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8366b9d773e11f211ad9bdbca5b34a6a38fde07","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-08-29 22:44:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"843e3c1746a9fe5ec47faa55bddf2bef858335d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-29 23:57:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/25/597725/1/check/puppet-openstack-lint/15f6330/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-4 http://logs.openstack.org/25/597725/1/check/puppet-openstack-syntax-4/3a4d62f/ : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5 http://logs.openstack.org/25/597725/1/check/puppet-openstack-syntax-5/4dc9a9d/ : SUCCESS in 3m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-unit-4.8-centos-7/cd62c23/ : SUCCESS in 10m 43s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-unit-5.5-centos-7/ac36705/ : SUCCESS in 10m 59s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-unit-latest-centos-7/61abacb/ : SUCCESS in 11m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-beaker-centos-7/dca4676/ : SUCCESS in 25m 20s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/25/597725/1/check/puppet-openstack-beaker-ubuntu-bionic/b9d2b7b/ : FAILURE in 6m 59s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/25/597725/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/7b2cc32/ : SUCCESS in 45m 39s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/25/597725/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/151e4db/ : SUCCESS in 1h 08m 01s\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/25/597725/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/5ba1299/ : FAILURE in 38m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"163b29a27d5e86a644d7228bf5c9807749367f23","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-08-30 07:07:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f26f6fcf69da6fe9cc5106ef0c73050b4d6e5d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-30 08:18:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/25/597725/1/check/puppet-openstack-lint/54180b9/ : SUCCESS in 4m 00s\n- puppet-openstack-syntax-4 http://logs.openstack.org/25/597725/1/check/puppet-openstack-syntax-4/aa4a49b/ : SUCCESS in 3m 42s\n- puppet-openstack-syntax-5 http://logs.openstack.org/25/597725/1/check/puppet-openstack-syntax-5/38e32d5/ : SUCCESS in 3m 34s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-unit-4.8-centos-7/faf6bd7/ : SUCCESS in 16m 52s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-unit-5.5-centos-7/6d5eb99/ : SUCCESS in 10m 35s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-unit-latest-centos-7/62c8254/ : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/25/597725/1/check/puppet-openstack-beaker-centos-7/ded61d6/ : SUCCESS in 24m 15s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/25/597725/1/check/puppet-openstack-beaker-ubuntu-bionic/dec927b/ : SUCCESS in 15m 30s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/25/597725/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/99b867b/ : SUCCESS in 45m 56s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/25/597725/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/8e6a53b/ : SUCCESS in 1h 04m 28s\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/25/597725/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/4eba78e/ : FAILURE in 54m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddd961636d843196ae5baeda2212a67bd1760ea5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-08-30 18:05:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7b644f5df225060daa6c4376c91d21950b2a987","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-08-30 18:05:40.000000000","message":"Patch Set 1: Code-Review-1\n\nplease depends-on https://review.openstack.org/#/c/597600/","accounts_in_message":[],"_revision_number":1},{"id":"42c93c08cd4b6bd6c353b296b40cc9d89f54f43f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-08-30 18:12:13.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"09c5947e47fcdefcc50b42f631531d2df53d5602","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-30 19:21:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/25/597725/2/check/puppet-openstack-lint/b0ba29f/ : SUCCESS in 4m 31s\n- puppet-openstack-syntax-4 http://logs.openstack.org/25/597725/2/check/puppet-openstack-syntax-4/899af0c/ : SUCCESS in 4m 21s\n- puppet-openstack-syntax-5 http://logs.openstack.org/25/597725/2/check/puppet-openstack-syntax-5/fc00598/ : SUCCESS in 4m 29s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/25/597725/2/check/puppet-openstack-unit-4.8-centos-7/390a921/ : SUCCESS in 10m 56s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/25/597725/2/check/puppet-openstack-unit-5.5-centos-7/842579e/ : SUCCESS in 17m 14s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/25/597725/2/check/puppet-openstack-unit-latest-centos-7/6cf98ae/ : SUCCESS in 16m 10s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/25/597725/2/check/puppet-openstack-beaker-centos-7/83cbfad/ : SUCCESS in 24m 42s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/25/597725/2/check/puppet-openstack-beaker-ubuntu-bionic/5a89930/ : SUCCESS in 12m 16s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/25/597725/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/0b2f734/ : SUCCESS in 50m 51s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/25/597725/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/fb312f8/ : SUCCESS in 1h 04m 10s\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/25/597725/2/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/dfb20db/ : FAILURE in 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44ebcb538a46e9909e7c1021c1b3fd79de4380c9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-09-06 21:19:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4e10f016e7adeb8401c1b1050e9211a15d1eb0ee","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-01-02 10:07:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"877b4318e04a11999b716fca59e237a56ba05a22","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-01-02 14:25:16.000000000","message":"Patch Set 2:\n\nneeds a rebase","accounts_in_message":[],"_revision_number":2},{"id":"bd412f5f99d3f51ba1105aeb262e2f09703d9bd0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-24 22:28:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf5daffb9e9cc84d6e5c5976b227cab8d9243022","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 00:22:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/25/597725/3/check/puppet-openstack-lint-ubuntu-bionic/83e2aed/ : SUCCESS in 4m 28s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/25/597725/3/check/puppet-openstack-syntax-4-ubuntu-bionic/53161ba/ : SUCCESS in 3m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/25/597725/3/check/puppet-openstack-syntax-5-ubuntu-bionic/91ec9dc/ : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/25/597725/3/check/puppet-openstack-syntax-6-ubuntu-bionic/ef76025/ : SUCCESS in 3m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/25/597725/3/check/puppet-openstack-unit-4.8-centos-7/d618d49/ : SUCCESS in 11m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/25/597725/3/check/puppet-openstack-unit-5.5-centos-7/a940d69/ : SUCCESS in 10m 13s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/25/597725/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/5259823/ : SUCCESS in 13m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/25/597725/3/check/puppet-openstack-unit-latest-ubuntu-bionic/d603495/ : FAILURE in 8m 21s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/25/597725/3/check/puppet-openstack-beaker-centos-7/6f745d7/ : SUCCESS in 19m 26s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/25/597725/3/check/puppet-openstack-beaker-ubuntu-bionic/e55fc0a/ : FAILURE in 11m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/25/597725/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/66ea85c/ : FAILURE in 18m 04s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/25/597725/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/6799df1/ : FAILURE in 20m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9c54593600f173a01c9980fa36b2b707b8ced94a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-27 09:30:21.000000000","message":"Topic changed from 597725 to puppet-octavia-testing","accounts_in_message":[],"_revision_number":3},{"id":"bd6ff808577025515fcb69416b1a70efa9720143","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-10-11 09:01:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"82e1c1757c5df8771ddaa29ebeb2abb8a8706f61","revisions":{"d29e73a829c622211faff8b710df9c2b036532ff":{"kind":"REWORK","_number":1,"created":"2018-08-29 22:44:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/25/597725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/25/597725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/1"}}},"commit":{"parents":[{"commit":"54c9e53a27f51ee29bee1651aef6c1d16bd7da4c","subject":"Update reno for stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/54c9e53a27f51ee29bee1651aef6c1d16bd7da4c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-29 22:44:16.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-29 22:44:16.000000000","tz":120},"subject":"Add scenario001 to zuul config","message":"Add scenario001 to zuul config\n\nChange-Id: Icd17bfb5addf37d11207d008bd59583f929cb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d29e73a829c622211faff8b710df9c2b036532ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d29e73a829c622211faff8b710df9c2b036532ff"}]},"branch":"refs/heads/master"},"6b1dbb9334ff6bec6186bc4d902e9dfee656ecb7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-08-30 18:12:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/25/597725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/25/597725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/2"}}},"commit":{"parents":[{"commit":"54c9e53a27f51ee29bee1651aef6c1d16bd7da4c","subject":"Update reno for stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/54c9e53a27f51ee29bee1651aef6c1d16bd7da4c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-29 22:44:16.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-30 18:12:11.000000000","tz":0},"subject":"Add scenario001 to zuul config","message":"Add scenario001 to zuul config\n\nDepends-On: https://review.openstack.org/#/c/597600/\nChange-Id: Icd17bfb5addf37d11207d008bd59583f929cb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6b1dbb9334ff6bec6186bc4d902e9dfee656ecb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6b1dbb9334ff6bec6186bc4d902e9dfee656ecb7"}]},"branch":"refs/heads/master"},"82e1c1757c5df8771ddaa29ebeb2abb8a8706f61":{"kind":"REWORK","_number":3,"created":"2019-04-24 22:28:16.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/25/597725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/25/597725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/25/597725/3"}}},"commit":{"parents":[{"commit":"7c8cb7ec78f9d51bf8f67dc85b2bbc0cbd089539","subject":"Move to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7c8cb7ec78f9d51bf8f67dc85b2bbc0cbd089539"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-29 22:44:16.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-04-24 22:28:08.000000000","tz":120},"subject":"Add scenario001 to zuul config","message":"Add scenario001 to zuul config\n\nDepends-On: https://review.openstack.org/#/c/597600/\nChange-Id: Icd17bfb5addf37d11207d008bd59583f929cb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/82e1c1757c5df8771ddaa29ebeb2abb8a8706f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/82e1c1757c5df8771ddaa29ebeb2abb8a8706f61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
