)]}'
{"id":"openstack%2Fpuppet-octavia~655105","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fstein~I99cf987b8e40a6c79b699ff38631aa176f60cedb","project":"openstack/puppet-octavia","branch":"stable/stein","topic":"bug/1825146","hashtags":[],"change_id":"I99cf987b8e40a6c79b699ff38631aa176f60cedb","subject":"[stable/stein only] Remove validation for provider_drivers","status":"MERGED","created":"2019-04-23 12:01:59.000000000","updated":"2019-04-23 17:48:45.000000000","submitted":"2019-04-23 17:48:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655105-1556041725362-eaa21e4c","meta_rev_id":"662950d5ef76ea4ac5ab86e4a562fbd98ffa142f","_number":655105,"virtual_id_number":655105,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-23 17:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-04-23 15:20:58.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-04-23 17:00:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-04-23 17:00:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 15:20:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-23 17:00:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-23 17:48:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ff316fb13c36f72b201c410618fd56bcf229761","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-04-23 12:01:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"add0865d82316224bbe9d7a1522645d8abde4422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 12:49:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/05/655105/1/check/puppet-openstack-lint-ubuntu-bionic/bca829a/ : SUCCESS in 6m 55s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/05/655105/1/check/puppet-openstack-syntax-4-ubuntu-bionic/6c4382b/ : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/05/655105/1/check/puppet-openstack-syntax-5-ubuntu-bionic/bee5c4e/ : SUCCESS in 7m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/05/655105/1/check/puppet-openstack-syntax-6-ubuntu-bionic/d438019/ : SUCCESS in 5m 47s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/05/655105/1/check/puppet-openstack-unit-4.8-centos-7/dd1add6/ : SUCCESS in 20m 53s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/05/655105/1/check/puppet-openstack-unit-5.5-centos-7/580c82a/ : SUCCESS in 19m 38s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/05/655105/1/check/puppet-openstack-beaker-centos-7/2d58d2a/ : SUCCESS in 22m 45s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/05/655105/1/check/puppet-openstack-beaker-ubuntu-bionic/39460fd/ : FAILURE in 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1831547d483ee0d45ca3d22f7e848f9a62dc6c48","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-23 15:20:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71d3a5da66d50c27f9e641c7d8d9d79efbac1bd3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-23 17:00:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4895437422ccc05f1b0caf451f8e5e58759a39a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 17:00:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b85c15b0806b7917e2a32cd78e50218995557737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 17:48:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"70555d2216b743a92e872f8fc50a5ef4e071d121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 17:48:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/05/655105/1/gate/puppet-openstack-lint-ubuntu-bionic/373d73f/ : SUCCESS in 5m 07s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/05/655105/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/566aed1/ : SUCCESS in 3m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/05/655105/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/65a3ba2/ : SUCCESS in 8m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/05/655105/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/ae9d2ce/ : SUCCESS in 6m 28s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/05/655105/1/gate/puppet-openstack-unit-4.8-centos-7/774d075/ : SUCCESS in 20m 01s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/05/655105/1/gate/puppet-openstack-unit-5.5-centos-7/8cec36b/ : SUCCESS in 19m 17s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/05/655105/1/gate/puppet-openstack-beaker-centos-7/da3b374/ : SUCCESS in 24m 48s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/05/655105/1/gate/puppet-openstack-beaker-ubuntu-bionic/2d11bd5/ : FAILURE in 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e11a90e8d8e7fcd987225d06ee632fc9d92db8c","revisions":{"7e11a90e8d8e7fcd987225d06ee632fc9d92db8c":{"kind":"REWORK","_number":1,"created":"2019-04-23 12:01:59.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/05/655105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/05/655105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/05/655105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/05/655105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/05/655105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/05/655105/1"}}},"commit":{"parents":[{"commit":"02184a9927e5233f73aa7187df7b28b27ccb749c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/02184a9927e5233f73aa7187df7b28b27ccb749c"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-04-23 11:55:50.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-04-23 12:01:39.000000000","tz":330},"subject":"[stable/stein only] Remove validation for provider_drivers","message":"[stable/stein only] Remove validation for provider_drivers\n\nThis has been removed from master with commit\n119f263ea73bb354e706f3d479be6313358df85b.\n\nChange-Id: I99cf987b8e40a6c79b699ff38631aa176f60cedb\nRelated-Bug: #1825146\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7e11a90e8d8e7fcd987225d06ee632fc9d92db8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7e11a90e8d8e7fcd987225d06ee632fc9d92db8c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
