)]}'
{"id":"openstack%2Fpuppet-octavia~656126","triplet_id":"openstack%2Fpuppet-octavia~master~Ie628b4900fd6614d64498d78e9f948e874d9e953","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1826692","hashtags":[],"change_id":"Ie628b4900fd6614d64498d78e9f948e874d9e953","subject":"Deprecate idle_timeout option","status":"MERGED","created":"2019-04-28 07:00:56.000000000","updated":"2019-04-29 07:53:34.000000000","submitted":"2019-04-29 07:53:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656126-1556524414015-3a5dd909","meta_rev_id":"3e3d44640446fea14e36e30be5a38ab2cb29e6df","_number":656126,"virtual_id_number":656126,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-04-29 07:06:20.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-29 07:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-04-28 08:42:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-28 17:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-04-28 17:12:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 08:42:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-28 17:12:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-29 07:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"569a7bc387f69ea1147299e06a71024fe4940cf3","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 07:00:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ac58d581ef8406cb2f1afa1dfdb79ec846b9356","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 07:30:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/656126/1/check/puppet-openstack-lint-ubuntu-bionic/b5b666f/ : FAILURE in 5m 21s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/26/656126/1/check/puppet-openstack-syntax-4-ubuntu-bionic/9f08b13/ : SUCCESS in 3m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/656126/1/check/puppet-openstack-syntax-5-ubuntu-bionic/0834765/ : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/656126/1/check/puppet-openstack-syntax-6-ubuntu-bionic/1b7226c/ : SUCCESS in 4m 13s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/26/656126/1/check/puppet-openstack-unit-4.8-centos-7/7dd0c20/ : SUCCESS in 14m 20s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/656126/1/check/puppet-openstack-unit-5.5-centos-7/65bf96d/ : SUCCESS in 14m 07s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/656126/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/625292b/ : SUCCESS in 10m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/656126/1/check/puppet-openstack-unit-latest-ubuntu-bionic/4ebd804/ : SUCCESS in 10m 59s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/26/656126/1/check/puppet-openstack-beaker-centos-7/a09352e/ : SUCCESS in 23m 01s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/26/656126/1/check/puppet-openstack-beaker-ubuntu-bionic/985c5c1/ : FAILURE in 7m 20s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/26/656126/1/check/build-openstack-releasenotes/dbd2ead/html/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cbbbb2d989eacd209c22337291379ac18058a185","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 07:39:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"980c1a3c66a95c00649ccbfd9d0486eec9c9e0f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 08:04:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/656126/2/check/puppet-openstack-lint-ubuntu-bionic/0083875/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/26/656126/2/check/puppet-openstack-syntax-4-ubuntu-bionic/e44727c/ : FAILURE in 3m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/656126/2/check/puppet-openstack-syntax-5-ubuntu-bionic/5d63b40/ : FAILURE in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/656126/2/check/puppet-openstack-syntax-6-ubuntu-bionic/9184d8e/ : FAILURE in 3m 36s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/26/656126/2/check/puppet-openstack-unit-4.8-centos-7/2b9d7c0/ : FAILURE in 12m 13s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/656126/2/check/puppet-openstack-unit-5.5-centos-7/d7ae7e9/ : FAILURE in 12m 59s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/656126/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/3bdaf3d/ : FAILURE in 7m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/656126/2/check/puppet-openstack-unit-latest-ubuntu-bionic/02676a1/ : FAILURE in 6m 23s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/26/656126/2/check/puppet-openstack-beaker-centos-7/f3d9c5e/ : FAILURE in 9m 37s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/26/656126/2/check/puppet-openstack-beaker-ubuntu-bionic/9d819d9/ : FAILURE in 7m 22s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/26/656126/2/check/build-openstack-releasenotes/af1d509/html/ : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"344b2b68770f12e484f26eae293a1d2a3aa57f12","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 08:08:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2c56ec438de1fcddf72cad38f5e9258dac121f4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 08:29:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"89ef83d63d16791895fcbe1b8ccd8be6e9beaa07","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-28 08:42:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cf08363538fc3adf852ee777e3a854b7be05c0bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 08:48:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-lint-ubuntu-bionic/7db4483/ : SUCCESS in 3m 53s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-syntax-4-ubuntu-bionic/08b2087/ : SUCCESS in 3m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-syntax-5-ubuntu-bionic/6439a2e/ : SUCCESS in 6m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-syntax-6-ubuntu-bionic/920e5a1/ : SUCCESS in 4m 22s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-4.8-centos-7/64ed3dc/ : SUCCESS in 11m 34s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-5.5-centos-7/1fff516/ : SUCCESS in 10m 36s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/a941aaf/ : SUCCESS in 8m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-latest-ubuntu-bionic/8f451e6/ : SUCCESS in 7m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/26/656126/4/check/puppet-openstack-beaker-centos-7/26673ff/ : SUCCESS in 19m 02s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-beaker-ubuntu-bionic/ac8bacc/ : FAILURE in 9m 14s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/26/656126/4/check/build-openstack-releasenotes/e707836/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":4},{"id":"46f1136e6dfb856a6952392b6b77f013fed0740d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-28 17:12:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"055c02373337cbac5ad61f641f0d1ec151b24ef8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-29 07:06:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f294e85ecac7acc693fe8d69828bd5569536e4c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:31:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-lint-ubuntu-bionic/fe8b0da/ : SUCCESS in 3m 51s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-syntax-4-ubuntu-bionic/ca23270/ : SUCCESS in 4m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-syntax-5-ubuntu-bionic/729d7d6/ : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-syntax-6-ubuntu-bionic/cddf054/ : SUCCESS in 3m 41s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-4.8-centos-7/b0f60b5/ : SUCCESS in 11m 54s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-5.5-centos-7/08893d8/ : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/c3eeef7/ : SUCCESS in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-unit-latest-ubuntu-bionic/5c50aa2/ : SUCCESS in 8m 40s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/26/656126/4/check/puppet-openstack-beaker-centos-7/852e473/ : SUCCESS in 19m 33s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/26/656126/4/check/puppet-openstack-beaker-ubuntu-bionic/25b1124/ : FAILURE in 7m 26s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/26/656126/4/check/build-openstack-releasenotes/db00088/html/ : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":4},{"id":"16f9407d987c901d4b13a950c5eec253bfd4a216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:31:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0081a44915ebd9b84b0eb1267f2d827ca639ea9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:53:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/656126/4/gate/puppet-openstack-lint-ubuntu-bionic/c4ec6e7/ : SUCCESS in 4m 09s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/26/656126/4/gate/puppet-openstack-syntax-4-ubuntu-bionic/7f92147/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/656126/4/gate/puppet-openstack-syntax-5-ubuntu-bionic/5c39193/ : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/656126/4/gate/puppet-openstack-syntax-6-ubuntu-bionic/afb6967/ : SUCCESS in 4m 26s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/26/656126/4/gate/puppet-openstack-unit-4.8-centos-7/5bb6200/ : SUCCESS in 11m 49s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/656126/4/gate/puppet-openstack-unit-5.5-centos-7/e666fcf/ : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/656126/4/gate/puppet-openstack-unit-6.0-ubuntu-bionic/f047048/ : SUCCESS in 12m 39s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/26/656126/4/gate/puppet-openstack-beaker-centos-7/ad62fda/ : SUCCESS in 19m 52s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/26/656126/4/gate/puppet-openstack-beaker-ubuntu-bionic/b766aef/ : FAILURE in 11m 11s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/26/656126/4/gate/build-openstack-releasenotes/136abd1/html/ : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":4},{"id":"567da0e29d3ac4ce0a9e992116ba199fc0f07f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:53:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a4f8833b726df2ebe4a84810b10262b0191405f5","revisions":{"72e2e0118d738accdf3278b47d61f38b08f95f27":{"kind":"REWORK","_number":1,"created":"2019-04-28 07:00:56.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/26/656126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/656126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/1"}}},"commit":{"parents":[{"commit":"29805c10de7c873ecea4bd90f14c285ee5e73af4","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/29805c10de7c873ecea4bd90f14c285ee5e73af4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:11.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:11.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: Ie628b4900fd6614d64498d78e9f948e874d9e953\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/72e2e0118d738accdf3278b47d61f38b08f95f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/72e2e0118d738accdf3278b47d61f38b08f95f27"}]},"branch":"refs/heads/master"},"a3c12ee642f1a0fbf2008849e48a29f7445c6740":{"kind":"REWORK","_number":2,"created":"2019-04-28 07:39:12.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/26/656126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/656126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/2"}}},"commit":{"parents":[{"commit":"29805c10de7c873ecea4bd90f14c285ee5e73af4","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/29805c10de7c873ecea4bd90f14c285ee5e73af4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:11.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:38:55.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: Ie628b4900fd6614d64498d78e9f948e874d9e953\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c12ee642f1a0fbf2008849e48a29f7445c6740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c12ee642f1a0fbf2008849e48a29f7445c6740"}]},"branch":"refs/heads/master"},"a392f450efce8ef321438cf7fff73c941f2844a5":{"kind":"REWORK","_number":3,"created":"2019-04-28 08:08:27.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/26/656126/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/656126/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/3"}}},"commit":{"parents":[{"commit":"29805c10de7c873ecea4bd90f14c285ee5e73af4","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/29805c10de7c873ecea4bd90f14c285ee5e73af4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:11.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 08:07:52.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: Ie628b4900fd6614d64498d78e9f948e874d9e953\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a392f450efce8ef321438cf7fff73c941f2844a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a392f450efce8ef321438cf7fff73c941f2844a5"}]},"branch":"refs/heads/master"},"a4f8833b726df2ebe4a84810b10262b0191405f5":{"kind":"REWORK","_number":4,"created":"2019-04-28 08:29:13.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/26/656126/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/656126/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/656126/4"}}},"commit":{"parents":[{"commit":"29805c10de7c873ecea4bd90f14c285ee5e73af4","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/29805c10de7c873ecea4bd90f14c285ee5e73af4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:11.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 08:28:58.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: Ie628b4900fd6614d64498d78e9f948e874d9e953\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a4f8833b726df2ebe4a84810b10262b0191405f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a4f8833b726df2ebe4a84810b10262b0191405f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
