)]}'
{"id":"openstack%2Fpuppet-octavia~659407","triplet_id":"openstack%2Fpuppet-octavia~master~I46eac78b8213cb5c3b61e2ae9e73c30d5fd420a1","project":"openstack/puppet-octavia","branch":"master","topic":"add_openstackclient","hashtags":[],"change_id":"I46eac78b8213cb5c3b61e2ae9e73c30d5fd420a1","subject":"Add openstackclient installation to the client class","status":"MERGED","created":"2019-05-16 01:53:06.000000000","updated":"2019-05-17 19:53:56.000000000","submitted":"2019-05-17 19:53:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659407-1558122836314-38c90c1d","meta_rev_id":"3725c9d18e422426a939d166662ba2dc6a4f193a","_number":659407,"virtual_id_number":659407,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-17 19:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-05-17 09:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-05-17 19:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-05-17 19:28:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 09:55:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-05-17 19:28:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-17 19:53:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffc3d9d31bb31bd14a0177f458506a7ee4096a3e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-16 01:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eed2916a355b74eb975f8115f9aef9c20dbc7356","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 02:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/07/659407/1/check/puppet-openstack-lint-ubuntu-bionic/cbd05d2/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/07/659407/1/check/puppet-openstack-syntax-4-ubuntu-bionic/59157aa/ : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/07/659407/1/check/puppet-openstack-syntax-5-ubuntu-bionic/f018b74/ : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/07/659407/1/check/puppet-openstack-syntax-6-ubuntu-bionic/75ab40d/ : SUCCESS in 4m 05s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/659407/1/check/puppet-openstack-unit-4.8-centos-7/9321392/ : SUCCESS in 25m 01s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/659407/1/check/puppet-openstack-unit-5.5-centos-7/64e6af1/ : SUCCESS in 23m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/07/659407/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/3b90185/ : SUCCESS in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/07/659407/1/check/puppet-openstack-unit-latest-ubuntu-bionic/ada8c55/ : SUCCESS in 8m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/659407/1/check/puppet-openstack-beaker-centos-7/e1fc496/ : SUCCESS in 37m 14s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/07/659407/1/check/puppet-openstack-beaker-ubuntu-bionic/b379f66/ : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"125daf64f2e0af8c5bfcb522c629c30f152b867a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 01:01:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"015da090e7491d52af7d23a560597e6f1d62a111","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 01:27:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/07/659407/2/check/puppet-openstack-lint-ubuntu-bionic/6459208/ : SUCCESS in 6m 28s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/07/659407/2/check/puppet-openstack-syntax-4-ubuntu-bionic/1126aac/ : SUCCESS in 6m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/07/659407/2/check/puppet-openstack-syntax-5-ubuntu-bionic/74c3149/ : SUCCESS in 6m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/07/659407/2/check/puppet-openstack-syntax-6-ubuntu-bionic/267ea54/ : SUCCESS in 6m 16s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/659407/2/check/puppet-openstack-unit-4.8-centos-7/45b0f9d/ : SUCCESS in 17m 53s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/659407/2/check/puppet-openstack-unit-5.5-centos-7/5c7b806/ : SUCCESS in 17m 08s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/07/659407/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/523f5fa/ : SUCCESS in 10m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/07/659407/2/check/puppet-openstack-unit-latest-ubuntu-bionic/b08b830/ : SUCCESS in 10m 03s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/659407/2/check/puppet-openstack-beaker-centos-7/2c691a6/ : SUCCESS in 20m 12s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/07/659407/2/check/puppet-openstack-beaker-ubuntu-bionic/4af56e5/ : FAILURE in 9m 27s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/07/659407/2/check/build-openstack-releasenotes/1f3a4fe/html/ : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":2},{"id":"06f356f4f7ff292b3642d38fb5800029cb2f74ee","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 08:35:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"56e55501768d983cf801424bd981d347c6a4d6a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:08:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/07/659407/3/check/puppet-openstack-lint-ubuntu-bionic/6e6967b/ : SUCCESS in 3m 56s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/07/659407/3/check/puppet-openstack-syntax-4-ubuntu-bionic/8a505fe/ : SUCCESS in 6m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/07/659407/3/check/puppet-openstack-syntax-5-ubuntu-bionic/50f51f5/ : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/07/659407/3/check/puppet-openstack-syntax-6-ubuntu-bionic/338021e/ : SUCCESS in 6m 16s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/659407/3/check/puppet-openstack-unit-4.8-centos-7/96d5f1e/ : SUCCESS in 19m 54s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/659407/3/check/puppet-openstack-unit-5.5-centos-7/892871f/ : SUCCESS in 17m 05s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/07/659407/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/7121e9a/ : SUCCESS in 9m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/07/659407/3/check/puppet-openstack-unit-latest-ubuntu-bionic/c6d3be0/ : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/659407/3/check/puppet-openstack-beaker-centos-7/dcfabe1/ : SUCCESS in 20m 10s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/07/659407/3/check/puppet-openstack-beaker-ubuntu-bionic/4ae18db/ : FAILURE in 7m 04s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/07/659407/3/check/build-openstack-releasenotes/60f6810/html/ : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f83845943ec9201f57c99cb73a3181c138b44b2f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-17 09:55:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a2de18e1aeea53b598f576ecce37b9b9a754fc54","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-17 19:28:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"795239a95ffbf3b4197463d9596bfd44eb87aea4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:28:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1f1d345a2489ff00aee7efcb7a9a40aa2e2231c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:53:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/07/659407/3/gate/puppet-openstack-lint-ubuntu-bionic/69cf036/ : SUCCESS in 6m 27s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/07/659407/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/9aa473f/ : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/07/659407/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/cc220de/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/07/659407/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/a11e864/ : SUCCESS in 3m 39s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/659407/3/gate/puppet-openstack-unit-4.8-centos-7/3562e33/ : SUCCESS in 18m 18s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/659407/3/gate/puppet-openstack-unit-5.5-centos-7/dcdcc3c/ : SUCCESS in 17m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/07/659407/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/3e6e578/ : SUCCESS in 9m 05s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/659407/3/gate/puppet-openstack-beaker-centos-7/7b5302c/ : SUCCESS in 19m 38s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/07/659407/3/gate/puppet-openstack-beaker-ubuntu-bionic/eddfd03/ : FAILURE in 8m 07s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/07/659407/3/gate/build-openstack-releasenotes/239b91b/html/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"3d3deb942e48aba0f7dbdafc4097b85445a8423b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:53:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f6f355dcfe78411ba876022707fd1f05762944ca","revisions":{"f4f27b3de8460fb85f06fe58b96ae5f5a62a135c":{"kind":"REWORK","_number":1,"created":"2019-05-16 01:53:06.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/07/659407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/07/659407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/1"}}},"commit":{"parents":[{"commit":"a4f8833b726df2ebe4a84810b10262b0191405f5","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a4f8833b726df2ebe4a84810b10262b0191405f5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"subject":"Add openstackclient","message":"Add openstackclient\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I46eac78b8213cb5c3b61e2ae9e73c30d5fd420a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f4f27b3de8460fb85f06fe58b96ae5f5a62a135c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f4f27b3de8460fb85f06fe58b96ae5f5a62a135c"}]},"branch":"refs/heads/master"},"55bde31ed48aef4b4ec1f6df7a161310b1092f5d":{"kind":"REWORK","_number":2,"created":"2019-05-17 01:01:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/07/659407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/07/659407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/2"}}},"commit":{"parents":[{"commit":"a4f8833b726df2ebe4a84810b10262b0191405f5","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a4f8833b726df2ebe4a84810b10262b0191405f5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 01:00:27.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I46eac78b8213cb5c3b61e2ae9e73c30d5fd420a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/55bde31ed48aef4b4ec1f6df7a161310b1092f5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/55bde31ed48aef4b4ec1f6df7a161310b1092f5d"}]},"branch":"refs/heads/master"},"f6f355dcfe78411ba876022707fd1f05762944ca":{"kind":"REWORK","_number":3,"created":"2019-05-17 08:35:38.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/07/659407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/07/659407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/07/659407/3"}}},"commit":{"parents":[{"commit":"a4f8833b726df2ebe4a84810b10262b0191405f5","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a4f8833b726df2ebe4a84810b10262b0191405f5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 08:31:20.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I46eac78b8213cb5c3b61e2ae9e73c30d5fd420a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6f355dcfe78411ba876022707fd1f05762944ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6f355dcfe78411ba876022707fd1f05762944ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
