)]}'
{"id":"openstack%2Fpuppet-octavia~660007","triplet_id":"openstack%2Fpuppet-octavia~master~Ic7848e58b8ffeca624bdd6796b739a238505f5ef","project":"openstack/puppet-octavia","branch":"master","topic":"fix-client-test","hashtags":[],"change_id":"Ic7848e58b8ffeca624bdd6796b739a238505f5ef","subject":"Fix client spec test","status":"MERGED","created":"2019-05-20 03:05:01.000000000","updated":"2019-05-20 04:46:17.000000000","submitted":"2019-05-20 04:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660007-1558327577381-4b33e095","meta_rev_id":"79eb5191819d1d660ca1c4ba686ea2c0d3f2d557","_number":660007,"virtual_id_number":660007,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-20 04:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-20 04:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 04:17:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 04:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d105c75fc00361db3e0ee90c4fbcc8106f9cbaf","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 03:05:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8ed18d9382ad4ebd6997323c9d4e67c7d826b6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:29:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/07/660007/1/check/puppet-openstack-lint-ubuntu-bionic/4b9fd9c/ : SUCCESS in 3m 45s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/07/660007/1/check/puppet-openstack-syntax-4-ubuntu-bionic/43f236f/ : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/07/660007/1/check/puppet-openstack-syntax-5-ubuntu-bionic/c7ddc8c/ : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/07/660007/1/check/puppet-openstack-syntax-6-ubuntu-bionic/65e29c1/ : SUCCESS in 3m 30s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/660007/1/check/puppet-openstack-unit-4.8-centos-7/97051eb/ : SUCCESS in 11m 30s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/660007/1/check/puppet-openstack-unit-5.5-centos-7/3b4d649/ : SUCCESS in 10m 09s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/07/660007/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/e541b06/ : SUCCESS in 6m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/07/660007/1/check/puppet-openstack-unit-latest-ubuntu-bionic/ab644ff/ : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58add2b8b87d074e3d8818f7fdcad032dfb7cd15","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 04:17:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9214d73a0346395fbbadca5afc76b4a5e5d8ea0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:17:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c4f831c021a270eadf58eefddbb8e04cec0abeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:46:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eb21dfd433a3f6dd892e9f73f696aa3f0724088c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:46:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/07/660007/1/gate/puppet-openstack-lint-ubuntu-bionic/84c9147/ : SUCCESS in 6m 18s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/07/660007/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/cd92c0e/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/07/660007/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/396fb65/ : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/07/660007/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/9292a13/ : SUCCESS in 6m 12s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/660007/1/gate/puppet-openstack-unit-4.8-centos-7/a4fbe24/ : SUCCESS in 11m 29s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/660007/1/gate/puppet-openstack-unit-5.5-centos-7/ea6adb5/ : SUCCESS in 10m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/07/660007/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/279f9be/ : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","revisions":{"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e":{"kind":"REWORK","_number":1,"created":"2019-05-20 03:05:01.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/07/660007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/07/660007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/660007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/660007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/07/660007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/07/660007/1"}}},"commit":{"parents":[{"commit":"f6f355dcfe78411ba876022707fd1f05762944ca","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6f355dcfe78411ba876022707fd1f05762944ca"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:20.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:20.000000000","tz":480},"subject":"Fix client spec test","message":"Fix client spec test\n\nChange-Id: Ic7848e58b8ffeca624bdd6796b739a238505f5ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
