)]}'
{"id":"openstack%2Fpuppet-octavia~661952","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fqueens~Id6c0d156715147c6559dc39098a6eaabf77ac426","project":"openstack/puppet-octavia","branch":"stable/queens","topic":"server_certs_key_passphrase-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Id6c0d156715147c6559dc39098a6eaabf77ac426","subject":"Configure server_certs_key_passphrase for Octavia","status":"MERGED","created":"2019-05-29 12:00:21.000000000","updated":"2019-05-31 00:16:06.000000000","submitted":"2019-05-31 00:16:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661952-1559261766180-db54cbaf","meta_rev_id":"0430ee1c93190388138e1221829f79399c07b107","_number":661952,"virtual_id_number":661952,"owner":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-31 00:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-30 00:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-30 23:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-05-30 23:34:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 00:27:08.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-05-30 23:34:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-31 00:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07df10b16dd2f6eccd2201fc00179fc93905943d","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-05-29 12:00:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"4bf273f8ac835e774b1cd6dc39bb583a7787c32b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 12:51:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/52/661952/1/check/puppet-openstack-lint/e1bba31/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-4 http://logs.openstack.org/52/661952/1/check/puppet-openstack-syntax-4/f848e47/ : SUCCESS in 5m 02s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/52/661952/1/check/puppet-openstack-unit-4.8-centos-7/56e177a/ : SUCCESS in 12m 44s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/52/661952/1/check/puppet-openstack-unit-5.5-centos-7/ae0a4bb/ : SUCCESS in 12m 15s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/52/661952/1/check/puppet-openstack-beaker-centos-7/f3376c4/ : SUCCESS in 19m 39s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/52/661952/1/check/puppet-openstack-beaker-xenial/3219ae4/ : SUCCESS in 13m 36s\n- build-openstack-releasenotes http://logs.openstack.org/52/661952/1/check/build-openstack-releasenotes/fa91aef/html/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"37a2bfa1df2e1f2c365ce4e891a87701217e5e97","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-30 00:27:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5095fd8beb47d1cce1784c68f791afae3f26ea24","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-30 23:34:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4cbcb1d82259ad8600d4748529925a0863d8b86b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 23:34:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f405e4ce55c7c72b86b95449914ed3952ece0c01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 00:16:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/52/661952/1/gate/puppet-openstack-lint/1ecae73/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-4 http://logs.openstack.org/52/661952/1/gate/puppet-openstack-syntax-4/f8b5d09/ : SUCCESS in 4m 00s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/52/661952/1/gate/puppet-openstack-unit-4.8-centos-7/4369754/ : SUCCESS in 14m 30s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/52/661952/1/gate/puppet-openstack-unit-5.5-centos-7/e94c8c6/ : SUCCESS in 10m 23s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/52/661952/1/gate/puppet-openstack-beaker-centos-7/8d075c9/ : SUCCESS in 22m 15s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/52/661952/1/gate/puppet-openstack-beaker-xenial/7455684/ : SUCCESS in 10m 47s\n- build-openstack-releasenotes http://logs.openstack.org/52/661952/1/gate/build-openstack-releasenotes/3a48194/html/ : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"739a3447b0400a49810e706d3eeda64210a8762c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 00:16:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdcd8132a1ef4e10b4c21832ca95fe890ea6badd","revisions":{"bdcd8132a1ef4e10b4c21832ca95fe890ea6badd":{"kind":"REWORK","_number":1,"created":"2019-05-29 12:00:21.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/52/661952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/52/661952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/52/661952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/52/661952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/52/661952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/52/661952/1"}}},"commit":{"parents":[{"commit":"d83ceb023d1c4e940d746c531a87b77b8aeba0d7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d83ceb023d1c4e940d746c531a87b77b8aeba0d7"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-03-25 15:08:31.000000000","tz":120},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-05-29 12:00:21.000000000","tz":0},"subject":"Configure server_certs_key_passphrase for Octavia","message":"Configure server_certs_key_passphrase for Octavia\n\nA recent change[1] to Octavia added a parameter named\nserver_certs_key_passphrase, which means that TripleO should\ngenerate a password for it to avoid using the default value.\n\nCloses-Bug: #1821751\n\n[1] I06d329ca53bc36bd27f7870ae7c7ca0cf18575b2\n\nChange-Id: Id6c0d156715147c6559dc39098a6eaabf77ac426\n(cherry picked from commit e8ae4607b43b2b5f686f7337540837cc024c0b71)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bdcd8132a1ef4e10b4c21832ca95fe890ea6badd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bdcd8132a1ef4e10b4c21832ca95fe890ea6badd"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
