)]}'
{"id":"openstack%2Fpuppet-octavia~663028","triplet_id":"openstack%2Fpuppet-octavia~master~If719e37d3645079e2d78df105578a2c2cd8a7e9b","project":"openstack/puppet-octavia","branch":"master","topic":"remove-lbaas","hashtags":[],"change_id":"If719e37d3645079e2d78df105578a2c2cd8a7e9b","subject":"Deprecate event_streamer_driver parameter","status":"MERGED","created":"2019-06-04 13:23:52.000000000","updated":"2019-06-12 01:31:21.000000000","submitted":"2019-06-12 01:31:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663028-1560303081463-216c3fa0","meta_rev_id":"98540b27b9596d02642589d7f7a157c9c8c6ae56","_number":663028,"virtual_id_number":663028,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-06-12 01:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-12 00:39:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-06-07 13:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-06-12 00:39:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 17:37:13.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-06-05 01:15:01.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-06-07 13:53:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-06-12 00:39:29.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-12 01:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d2fe305bc73c084d5017752d48de03f7588edb0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-04 13:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46b8456032c30962874725b81b33bed520fa1537","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 13:58:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/28/663028/1/check/puppet-openstack-lint-ubuntu-bionic/8f586de/ : SUCCESS in 4m 13s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/28/663028/1/check/puppet-openstack-syntax-4-ubuntu-bionic/1bb486c/ : SUCCESS in 3m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/28/663028/1/check/puppet-openstack-syntax-5-ubuntu-bionic/35e649e/ : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/28/663028/1/check/puppet-openstack-syntax-6-ubuntu-bionic/0330c2c/ : SUCCESS in 4m 10s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/28/663028/1/check/puppet-openstack-unit-4.8-centos-7/9006cc6/ : SUCCESS in 16m 22s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/28/663028/1/check/puppet-openstack-unit-5.5-centos-7/db15d43/ : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/28/663028/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/f32cbf0/ : SUCCESS in 8m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/28/663028/1/check/puppet-openstack-unit-latest-ubuntu-bionic/9930b8f/ : SUCCESS in 7m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/28/663028/1/check/puppet-openstack-beaker-centos-7/3a454e6/ : SUCCESS in 19m 47s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/28/663028/1/check/puppet-openstack-beaker-ubuntu-bionic/12c973c/ : FAILURE in 7m 19s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/28/663028/1/check/build-openstack-releasenotes/48dcf96/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cb683a933afe6306bdf1cbdee0794e2ff99fdb7c","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-04 17:37:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"87e50462521799bb076f420c100f8afa9dd85baa","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:15:01.000000000","message":"Patch Set 1: Code-Review-1\n\nplease deprecate the parameter first, for example:\nhttps://review.opendev.org/#/c/592888/","accounts_in_message":[],"_revision_number":1},{"id":"09e92a9ca0acab59f64b58281db9995346a18bbf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-06 17:06:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bc8c5a6948d1159fc3c0e25bd2b69dacb833123","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 17:34:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/28/663028/2/check/puppet-openstack-lint-ubuntu-bionic/feb78bd/ : SUCCESS in 3m 42s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/28/663028/2/check/puppet-openstack-syntax-4-ubuntu-bionic/21a9bf7/ : SUCCESS in 6m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/28/663028/2/check/puppet-openstack-syntax-5-ubuntu-bionic/8c76fa2/ : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/28/663028/2/check/puppet-openstack-syntax-6-ubuntu-bionic/786e9e9/ : SUCCESS in 6m 33s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/28/663028/2/check/puppet-openstack-unit-4.8-centos-7/b1f667b/ : SUCCESS in 18m 40s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/28/663028/2/check/puppet-openstack-unit-5.5-centos-7/a5458a9/ : SUCCESS in 18m 29s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/28/663028/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/710a519/ : SUCCESS in 12m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/28/663028/2/check/puppet-openstack-unit-latest-ubuntu-bionic/87cea22/ : SUCCESS in 7m 59s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/28/663028/2/check/puppet-openstack-beaker-centos-7/a89c9ae/ : SUCCESS in 19m 43s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/28/663028/2/check/puppet-openstack-beaker-ubuntu-bionic/0287acd/ : FAILURE in 7m 22s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/28/663028/2/check/build-openstack-releasenotes/74d20d6/html/ : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6bca8700b47c253fc464d9f55f2fa7adc67a1a71","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-07 13:53:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b3a153fc5aa4a1006ad5927dd42ae8ae03140401","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 00:39:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a77dac7f18cafa18085e25aca567cd06e40ceaf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:39:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"987b97db3c1c9c507b865edb41b25764c56ef59b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 01:31:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"84eda712e1b0fa74cfbf1c203bd17a136a716720","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 01:31:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/28/663028/2/gate/puppet-openstack-lint-ubuntu-bionic/06cc44f/ : SUCCESS in 6m 47s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/28/663028/2/gate/puppet-openstack-syntax-4-ubuntu-bionic/d168116/ : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/28/663028/2/gate/puppet-openstack-syntax-5-ubuntu-bionic/87fd4be/ : SUCCESS in 6m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/28/663028/2/gate/puppet-openstack-syntax-6-ubuntu-bionic/77d5efb/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/28/663028/2/gate/puppet-openstack-unit-4.8-centos-7/e024e9d/ : SUCCESS in 12m 41s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/28/663028/2/gate/puppet-openstack-unit-5.5-centos-7/db8a7d1/ : SUCCESS in 11m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/28/663028/2/gate/puppet-openstack-unit-6.4-ubuntu-bionic/d2005bd/ : SUCCESS in 7m 26s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/28/663028/2/gate/puppet-openstack-beaker-centos-7/4eb37a6/ : SUCCESS in 20m 46s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/28/663028/2/gate/puppet-openstack-beaker-ubuntu-bionic/616b1f2/ : FAILURE in 8m 32s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/28/663028/2/gate/build-openstack-releasenotes/d27d1b7/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"748f26fe2282ea00f5d71b4af2335eac8273cd9b","revisions":{"95bd044551d933c02d6ebb34338e984ff7dcf35b":{"kind":"REWORK","_number":1,"created":"2019-06-04 13:23:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/663028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/28/663028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/1"}}},"commit":{"parents":[{"commit":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-04 13:22:09.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-04 13:23:38.000000000","tz":120},"subject":"Remove event_streamer_driver parameter","message":"Remove event_streamer_driver parameter\n\nThe parameter was only useful for synchronizing Octavia and Neutron\nLBaaS databases. With Neutron LBaaS project now retired, providing this\nparameter is no longer needed.\n\nChange-Id: If719e37d3645079e2d78df105578a2c2cd8a7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/95bd044551d933c02d6ebb34338e984ff7dcf35b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/95bd044551d933c02d6ebb34338e984ff7dcf35b"}]},"branch":"refs/heads/master"},"748f26fe2282ea00f5d71b4af2335eac8273cd9b":{"kind":"REWORK","_number":2,"created":"2019-06-06 17:06:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/663028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/28/663028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/28/663028/2"}}},"commit":{"parents":[{"commit":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-04 13:22:09.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-06 17:05:24.000000000","tz":120},"subject":"Deprecate event_streamer_driver parameter","message":"Deprecate event_streamer_driver parameter\n\nThe parameter was only useful for synchronizing Octavia and Neutron\nLBaaS databases. With Neutron LBaaS project now retired, providing this\nparameter is no longer needed.\n\nChange-Id: If719e37d3645079e2d78df105578a2c2cd8a7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/748f26fe2282ea00f5d71b4af2335eac8273cd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/748f26fe2282ea00f5d71b4af2335eac8273cd9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
