)]}'
{"id":"openstack%2Fpuppet-octavia~663204","triplet_id":"openstack%2Fpuppet-octavia~master~I38c63269b9a2dce009c4ce233622ace848ae41ea","project":"openstack/puppet-octavia","branch":"master","topic":"prep-release-train-1","hashtags":[],"change_id":"I38c63269b9a2dce009c4ce233622ace848ae41ea","subject":"Prepare Train M2","status":"MERGED","created":"2019-06-05 01:42:59.000000000","updated":"2019-06-10 13:32:39.000000000","submitted":"2019-06-10 13:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663204-1560173559808-7c5f8a7f","meta_rev_id":"841166b4ea0bb42e0bfe10190cfe1208bdbd27d2","_number":663204,"virtual_id_number":663204,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 13:32:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-06-10 13:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-06-10 13:21:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 13:21:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 13:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91e9828f72fbbd46a0a7ce384067d5a6264c55c9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9705527eac5738803ecfcd5aa1dd2efef065a634","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:13:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/04/663204/1/check/build-openstack-puppet-tarball/e110685/ : SUCCESS in 2m 08s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/04/663204/1/check/puppet-openstack-lint-ubuntu-bionic/38619ce/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"56cdb9ff20f6d7721a89ebc9e48e42d33a144e57","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-10 02:09:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1c0bbca7714412b08bf9fad5c539fcf75911e7bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:14:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/04/663204/2/check/build-openstack-puppet-tarball/2fa11f2/ : SUCCESS in 2m 10s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/04/663204/2/check/puppet-openstack-lint-ubuntu-bionic/36038aa/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ea9453ab0db2499e344dd21e5f46d4542a93a68d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:21:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aae6e3bc425b0c4d1152a60d42d932a03390d1c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:21:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"386f30dfdd9acfc0e4e4a5539a41299b718ecf6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:32:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fb86034b09746d4c73b791dbc7dd30f9c7a43032","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:32:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/04/663204/2/gate/build-openstack-puppet-tarball/4a2c12d/ : SUCCESS in 1m 51s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/04/663204/2/gate/puppet-openstack-lint-ubuntu-bionic/e5c286c/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7a171e21ceace57bf1e47b33bbdeac4363d3136","revisions":{"f5bcfa53f639fc394f5729dfbff420f5fb486dcf":{"kind":"REWORK","_number":1,"created":"2019-06-05 01:42:59.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/04/663204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/04/663204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/1"}}},"commit":{"parents":[{"commit":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:21.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:21.000000000","tz":480},"subject":"Prepare Train M1","message":"Prepare Train M1\n\nUpdate the version metadata for Train milestone 1\n\nChange-Id: I38c63269b9a2dce009c4ce233622ace848ae41ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f5bcfa53f639fc394f5729dfbff420f5fb486dcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f5bcfa53f639fc394f5729dfbff420f5fb486dcf"}]},"branch":"refs/heads/master"},"d7a171e21ceace57bf1e47b33bbdeac4363d3136":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 02:09:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/04/663204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/04/663204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/04/663204/2"}}},"commit":{"parents":[{"commit":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:21.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-10 02:04:44.000000000","tz":480},"subject":"Prepare Train M2","message":"Prepare Train M2\n\nUpdate the version metadata for Train milestone 2\n\nChange-Id: I38c63269b9a2dce009c4ce233622ace848ae41ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d7a171e21ceace57bf1e47b33bbdeac4363d3136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d7a171e21ceace57bf1e47b33bbdeac4363d3136"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
