)]}'
{"id":"openstack%2Fpuppet-octavia~663393","triplet_id":"openstack%2Fpuppet-octavia~master~Ica6f84256ff2068e9bac210e66a1429ea627256c","project":"openstack/puppet-octavia","branch":"master","hashtags":[],"change_id":"Ica6f84256ff2068e9bac210e66a1429ea627256c","subject":"Add driver agent configuration","status":"MERGED","created":"2019-06-05 18:10:37.000000000","updated":"2020-04-02 18:13:57.000000000","submitted":"2020-04-02 18:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663393-1585850751624-65716cc9","meta_rev_id":"d0a08f645cf382bd18e07a0a9840d54347a2a823","_number":663393,"virtual_id_number":663393,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 18:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-04-02 11:58:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-02 13:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-04-02 16:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-04-02 16:38:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 13:35:56.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-02 11:56:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-02 11:58:02.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-02 13:58:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-02 16:38:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-02 18:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96f0988c9e0415cda22968f2ba51f5b5ac85b7e6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-05 18:10:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc3c25b73482fb9f61a48a0ae43e55e0437bb314","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-05 18:10:51.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"58415236320dd5b5d012fcb8d0b336adb17e5a71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 18:55:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/93/663393/1/check/puppet-openstack-lint-ubuntu-bionic/f807fe7/ : FAILURE in 6m 26s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/93/663393/1/check/puppet-openstack-syntax-4-ubuntu-bionic/2646648/ : FAILURE in 6m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/93/663393/1/check/puppet-openstack-syntax-5-ubuntu-bionic/cd909e3/ : FAILURE in 6m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/93/663393/1/check/puppet-openstack-syntax-6-ubuntu-bionic/35b2a92/ : FAILURE in 3m 46s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/93/663393/1/check/puppet-openstack-unit-4.8-centos-7/72f3292/ : FAILURE in 9m 06s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/93/663393/1/check/puppet-openstack-unit-5.5-centos-7/c501e76/ : FAILURE in 9m 25s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/93/663393/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/630ada0/ : FAILURE in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/93/663393/1/check/puppet-openstack-unit-latest-ubuntu-bionic/86cf96f/ : FAILURE in 11m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/93/663393/1/check/puppet-openstack-beaker-centos-7/2058201/ : SUCCESS in 27m 35s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/93/663393/1/check/puppet-openstack-beaker-ubuntu-bionic/d20c618/ : FAILURE in 7m 14s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/93/663393/1/check/build-openstack-releasenotes/d0372b5/html/ : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"76d750c64f0d9de2390588bcc6ace942877b769d","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-05 18:57:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c93e7fcda1a4127cfc6e9e24688bf1448cf6a2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 19:56:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/93/663393/2/check/puppet-openstack-lint-ubuntu-bionic/34035ff/ : SUCCESS in 4m 21s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/93/663393/2/check/puppet-openstack-syntax-4-ubuntu-bionic/2040b99/ : FAILURE in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/93/663393/2/check/puppet-openstack-syntax-5-ubuntu-bionic/d97216c/ : FAILURE in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/93/663393/2/check/puppet-openstack-syntax-6-ubuntu-bionic/c40276b/ : FAILURE in 3m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/93/663393/2/check/puppet-openstack-unit-4.8-centos-7/92e6c73/ : FAILURE in 8m 32s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/93/663393/2/check/puppet-openstack-unit-5.5-centos-7/7246db6/ : FAILURE in 8m 57s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/93/663393/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/36d83f5/ : FAILURE in 6m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/93/663393/2/check/puppet-openstack-unit-latest-ubuntu-bionic/3e0bcfc/ : FAILURE in 8m 36s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/93/663393/2/check/puppet-openstack-beaker-centos-7/21f9e00/ : SUCCESS in 21m 59s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/93/663393/2/check/puppet-openstack-beaker-ubuntu-bionic/55389ef/ : FAILURE in 8m 23s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/93/663393/2/check/build-openstack-releasenotes/4ba0f7c/html/ : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"82dc5915eb1430f1d35581a8664297ef51b500bd","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-06 13:29:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc2936e639e3158c7374d978de0a7b2f9efaead3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 14:03:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/93/663393/3/check/puppet-openstack-lint-ubuntu-bionic/e0790be/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/93/663393/3/check/puppet-openstack-syntax-4-ubuntu-bionic/f8c9fde/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/93/663393/3/check/puppet-openstack-syntax-5-ubuntu-bionic/ee0931e/ : SUCCESS in 6m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/93/663393/3/check/puppet-openstack-syntax-6-ubuntu-bionic/3a431c6/ : SUCCESS in 6m 25s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/93/663393/3/check/puppet-openstack-unit-4.8-centos-7/8621baa/ : SUCCESS in 17m 31s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/93/663393/3/check/puppet-openstack-unit-5.5-centos-7/68e42c7/ : SUCCESS in 12m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/93/663393/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/a022c2c/ : SUCCESS in 8m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/93/663393/3/check/puppet-openstack-unit-latest-ubuntu-bionic/1f7b497/ : SUCCESS in 10m 15s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/93/663393/3/check/puppet-openstack-beaker-centos-7/8b4346b/ : SUCCESS in 20m 30s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/93/663393/3/check/puppet-openstack-beaker-ubuntu-bionic/e566e36/ : FAILURE in 8m 04s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/93/663393/3/check/build-openstack-releasenotes/af0da12/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"c9c9bf0db0466e1511896b3e02edc43f1cb9e0eb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-07 13:51:33.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1994a4d1bbdf7107948aa510815b49b654ff4ad3","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-07 13:53:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70775cecafb68c01934b83b407d4017d49dcd310","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-01 19:42:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"23b8e28a81a200dba18b196bdbed1e7fb8cfab3c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-01 19:43:46.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ddfdb6a327a016b6f477dbefe317bd17d9a6a2bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 20:18:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e6093a455b0454693e96b5c4ff072ae : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acf216a5271d41cd8c537af473b03a7c : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3bcee49a81e49d5a4357332c2c81769 : SUCCESS in 4m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/03e5ba6d8544464bb5420d8579cbc650 : FAILURE in 14m 15s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/cc60e1fde1784a2eaae9bb56118f4a41 : FAILURE in 11m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c32de8ae984d4976b7e9a2f5c0e82134 : FAILURE in 9m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/485f77a5e9134c89949ce6ef7d81467e : FAILURE in 11m 14s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6838d6485a254477bb366a27a384608d : SUCCESS in 33m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed9972ef585145e78a221494a82c07ec : FAILURE in 7m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75c2bf1e3c4047219ee77f8743fb435b : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"1eced5720dfcbe7838a127e3a11e9af27e4f5cfc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-01 20:22:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ed7b6fd036d0b8db76885699cf5261ae5862627b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 20:52:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a396409bd78c4c8cab12d17cc03430f9 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b671f1bd5304a8db2a55e7d3a60785d : SUCCESS in 5m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d65ab5e2cbb4e06b262230532856ee3 : SUCCESS in 6m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7fe0bf15cbb14b889d2aeea01ef41984 : SUCCESS in 12m 24s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1a4be8aea15b417ca559bdf42234c1fb : SUCCESS in 12m 13s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/623c6a6a0ccb4d69a34693969f451d63 : SUCCESS in 10m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ec9a59590484beebbb3027ef35f62b7 : SUCCESS in 13m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/07be517c801c410da986e60fa4276738 : SUCCESS in 26m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb90c8e6202944a48c34dc6df08c9bef : FAILURE in 10m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d48b5db4eea4040907173f76ac53eb1 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":5},{"id":"6f2891a4d4f15ef9e53d2c24ec29d12f594a64b1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-01 22:06:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f356f8db3602c882c794d4201d0e2cf21034434d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 06:39:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f8db40d28ba762fbf87eef1f9235ab796bb13c1d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 06:39:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f8c72274c1e2c74d9d38d02df5bc7193ff673b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 07:16:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/209285185cae49d592ce94d7a4f056cd : SUCCESS in 7m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e832e337403648cfbbc6aea70ac362d8 : SUCCESS in 7m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25eecedd385f47c29ed6504e5f4e7709 : SUCCESS in 7m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8e0f1da62ae749fda8719341d0288e58 : SUCCESS in 18m 09s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/88bbb0fece774cb4a2198de05681afec : SUCCESS in 18m 03s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/045c1ca57ff64732b5998d5dd0a92049 : SUCCESS in 13m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89c7c17be5344c559598c3a4619c7b83 : SUCCESS in 15m 51s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a0f33d41492c4126ae7bb60f41f092c2 : SUCCESS in 27m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1818d00920b4ddcace33be587f2e3b9 : FAILURE in 9m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31d4691c19594e14a66cddb32b989d08 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":6},{"id":"d000c6c9f05c99d0f3832300baca4b621702966f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:56:32.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4c10a1ba13c47b7f7bb8c8eedd569991ae76b880","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:57:04.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"fc4aa3da00a86df372f21ca2bf2d2920650ef657","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 11:58:02.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"666bf73f3e1407f2905b974fb3c9fcd418188bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:26:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ab51f43da834218aec6a38bc41dc52c : SUCCESS in 5m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0f88c3b960f42eb97bf8d14cedf24af : SUCCESS in 5m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c651e1d906d44a984c3d72a32f960e0 : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/62f478036b6f4f81bbb2dcd48b006b44 : SUCCESS in 17m 25s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/4c9a74acc2a84f12bc20ae05d54d335e : SUCCESS in 14m 20s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aaba90ae54442329634eef99a8ce45a : SUCCESS in 12m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85ebdee14da044b6a2ae43691acec9b5 : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/558bae5c6ad8464fb8a7c5cc0315ce30 : SUCCESS in 24m 54s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b52b0187228048f6aa38457ee39ed3fd : FAILURE in 7m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc475335d7eb469f8b2ce1ce92d7603b : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":7},{"id":"d71fce486bc1c0a5bf919ba79cefa31680cc3353","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-02 13:58:53.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks :)","accounts_in_message":[],"_revision_number":7},{"id":"c62795834efa7dc78f786ea44adcc3a204ee26a7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-02 16:38:46.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"abf853b23309b5ceae5649c293bb85685cd434bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:11:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"527a0ec09fc6e20ebb62b614778bb2944945c36b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:05:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"e2806922dd5451cf0ecd3e3b8476b41a78bf2cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:05:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5623c99d5bca498f8528855726dfa0a3 : SUCCESS in 7m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4611a3b030fd48e9b9d59f3cf76ffbe7 : SUCCESS in 6m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7f0c5256013465c949e4d661a04a405 : SUCCESS in 5m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3907e6f1b5c6454997999eac834c34c7 : SUCCESS in 17m 49s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/e3924a37826a48c5a36293d538add3c4 : SUCCESS in 17m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76a86cb0e70c42a5bd7e34842fc80b08 : SUCCESS in 13m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f14c6b54ae07463c85573718fdba48f4 : SUCCESS in 33m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a72f05b01ea742d795597a003e9c5c3b : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":7},{"id":"2f844bba8b9f5f76c94ad6f679e11bf6479257e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:09:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77286c319b5b4232af628cf2ef7e4415 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":7},{"id":"d0a08f645cf382bd18e07a0a9840d54347a2a823","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 18:13:57.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 397d5fb86384fa59f225ca367d15207c40965912","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"88ff210c3021583921f41532b190fe0c3bafca62","revisions":{"afa7d96395deac243a7468d7f36ee75e57721e10":{"kind":"REWORK","_number":1,"created":"2019-06-05 18:10:37.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/93/663393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/663393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/1"}}},"commit":{"parents":[{"commit":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:10:25.000000000","tz":0},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/afa7d96395deac243a7468d7f36ee75e57721e10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/afa7d96395deac243a7468d7f36ee75e57721e10"}]},"branch":"refs/heads/master"},"155b1fe31c735e50ba06ff59711638c570894ff6":{"kind":"REWORK","_number":2,"created":"2019-06-05 18:57:42.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/93/663393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/663393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/2"}}},"commit":{"parents":[{"commit":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:57:36.000000000","tz":0},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/155b1fe31c735e50ba06ff59711638c570894ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/155b1fe31c735e50ba06ff59711638c570894ff6"}]},"branch":"refs/heads/master"},"5cf285faf66e918c826b95319657edc17bdadb48":{"kind":"REWORK","_number":3,"created":"2019-06-06 13:29:54.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/93/663393/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/663393/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/3"}}},"commit":{"parents":[{"commit":"b0b8878faaf6ad4f4d8372c42a0c8433702bf74e","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0b8878faaf6ad4f4d8372c42a0c8433702bf74e"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-06 13:29:47.000000000","tz":-150},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5cf285faf66e918c826b95319657edc17bdadb48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5cf285faf66e918c826b95319657edc17bdadb48"}]},"branch":"refs/heads/master"},"264aa1b6b8297395b4ec7b21a968461a5ccea5c3":{"kind":"REWORK","_number":4,"created":"2020-04-01 19:42:48.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/93/663393/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/663393/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/4"}}},"commit":{"parents":[{"commit":"17c5182534328130931186eedd6ec934a8b1534e","subject":"Merge \"Add Octavia OVN Provider configuration (2 of 2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/17c5182534328130931186eedd6ec934a8b1534e"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-01 19:42:18.000000000","tz":120},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/264aa1b6b8297395b4ec7b21a968461a5ccea5c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/264aa1b6b8297395b4ec7b21a968461a5ccea5c3"}]},"branch":"refs/heads/master"},"54c8e659b570ebe3192c91190ff68ece76ea08f1":{"kind":"REWORK","_number":5,"created":"2020-04-01 20:22:35.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/93/663393/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/663393/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/5"}}},"commit":{"parents":[{"commit":"17c5182534328130931186eedd6ec934a8b1534e","subject":"Merge \"Add Octavia OVN Provider configuration (2 of 2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/17c5182534328130931186eedd6ec934a8b1534e"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-01 20:22:06.000000000","tz":120},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/54c8e659b570ebe3192c91190ff68ece76ea08f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/54c8e659b570ebe3192c91190ff68ece76ea08f1"}]},"branch":"refs/heads/master"},"5f93e78756298143050be0ef85a715a961df8bba":{"kind":"REWORK","_number":6,"created":"2020-04-02 06:39:35.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/93/663393/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/663393/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/6"}}},"commit":{"parents":[{"commit":"17c5182534328130931186eedd6ec934a8b1534e","subject":"Merge \"Add Octavia OVN Provider configuration (2 of 2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/17c5182534328130931186eedd6ec934a8b1534e"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-02 06:39:23.000000000","tz":120},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f93e78756298143050be0ef85a715a961df8bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f93e78756298143050be0ef85a715a961df8bba"}]},"branch":"refs/heads/master"},"88ff210c3021583921f41532b190fe0c3bafca62":{"kind":"REWORK","_number":7,"created":"2020-04-02 11:57:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/663393/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/663393/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/663393/7"}}},"commit":{"parents":[{"commit":"17c5182534328130931186eedd6ec934a8b1534e","subject":"Merge \"Add Octavia OVN Provider configuration (2 of 2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/17c5182534328130931186eedd6ec934a8b1534e"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:56:55.000000000","tz":0},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/88ff210c3021583921f41532b190fe0c3bafca62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/88ff210c3021583921f41532b190fe0c3bafca62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
