)]}'
{"id":"openstack%2Fpuppet-octavia~665803","triplet_id":"openstack%2Fpuppet-octavia~master~I2f67ac6d8607e28fcc1255fbd5cf6ef0cb927923","project":"openstack/puppet-octavia","branch":"master","topic":"665803","hashtags":[],"change_id":"I2f67ac6d8607e28fcc1255fbd5cf6ef0cb927923","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:27:27.000000000","updated":"2020-09-18 22:56:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1bba655fd49741281285d9eaea4cf8b0355f053e","_number":665803,"virtual_id_number":665803,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-18 00:23:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3133334766d66a5a8ad0d7b22cd033526679905","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:27:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"117d184aee2ae9a88e539c4b358dac242b3ad84e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:50:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5103f15aa336768aa27845c4c924a721bf9cb6ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:23:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/03/665803/2/check/puppet-openstack-lint-ubuntu-bionic/642bc0e/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/03/665803/2/check/puppet-openstack-syntax-5-ubuntu-bionic/395d381/ : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/03/665803/2/check/puppet-openstack-syntax-6-ubuntu-bionic/3db065e/ : SUCCESS in 4m 31s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/03/665803/2/check/puppet-openstack-unit-5.5-centos-7/139db17/ : FAILURE in 13m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/03/665803/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/d140e09/ : FAILURE in 9m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/03/665803/2/check/puppet-openstack-unit-latest-ubuntu-bionic/b85285f/ : FAILURE in 8m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/03/665803/2/check/puppet-openstack-beaker-centos-7/746b2f1/ : FAILURE in 10m 01s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/03/665803/2/check/puppet-openstack-beaker-ubuntu-bionic/0fc5852/ : FAILURE in 7m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd1ffd7889fe5fc58948de18a92777ce0ca3eca8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:56:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"984bb99e1f78ad2ce0e844a719843fb5a104086e","revisions":{"3dedc982a25e7df3d34f9e11e11f5a42d2d6ce0b":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:27:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/03/665803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/03/665803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/1"}}},"commit":{"parents":[{"commit":"ebf09d916670c421df8c174231f121c2c13b2504","subject":"Merge \"Deprecate event_streamer_driver parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ebf09d916670c421df8c174231f121c2c13b2504"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:27:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:27:19.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I2f67ac6d8607e28fcc1255fbd5cf6ef0cb927923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3dedc982a25e7df3d34f9e11e11f5a42d2d6ce0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3dedc982a25e7df3d34f9e11e11f5a42d2d6ce0b"}]},"branch":"refs/heads/master"},"984bb99e1f78ad2ce0e844a719843fb5a104086e":{"kind":"REWORK","_number":2,"created":"2019-06-17 23:50:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/03/665803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/03/665803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/03/665803/2"}}},"commit":{"parents":[{"commit":"ebf09d916670c421df8c174231f121c2c13b2504","subject":"Merge \"Deprecate event_streamer_driver parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ebf09d916670c421df8c174231f121c2c13b2504"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:27:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:49:55.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I2f67ac6d8607e28fcc1255fbd5cf6ef0cb927923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/984bb99e1f78ad2ce0e844a719843fb5a104086e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/984bb99e1f78ad2ce0e844a719843fb5a104086e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
