)]}'
{"id":"openstack%2Fpuppet-octavia~669653","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fstein~I5c2629d9e7700fe1dd6f915bc257b1f058e40617","project":"openstack/puppet-octavia","branch":"stable/stein","topic":"OctaviaServerCertsKeyPassphrase-32chars","hashtags":[],"change_id":"I5c2629d9e7700fe1dd6f915bc257b1f058e40617","subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","status":"MERGED","created":"2019-07-08 12:58:40.000000000","updated":"2019-07-10 18:29:28.000000000","submitted":"2019-07-10 18:29:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669653-1562783368157-31ef7618","meta_rev_id":"6f4544cf252542322edc0f9efa8749db14dc892c","_number":669653,"virtual_id_number":669653,"owner":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-07-10 18:29:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-07-10 14:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-07-10 17:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-10 17:56:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-07-10 17:40:37.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 14:51:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-10 17:17:58.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-07-10 17:56:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-10 18:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f422e03e8cbfce97bcf8074a26b52fd18762f2bb","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-08 12:58:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b1ce1813cf6ba0e4c7286d5f6f7b1e0bb3b523c","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-08 12:58:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"113fc5a0759bd0b287e39426ae01d743f568a69d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 13:30:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-lint-ubuntu-bionic/f904375/ : SUCCESS in 5m 50s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-syntax-4-ubuntu-bionic/6b9ae42/ : SUCCESS in 5m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-syntax-5-ubuntu-bionic/57a360d/ : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-syntax-6-ubuntu-bionic/3e4d043/ : SUCCESS in 5m 34s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/53/669653/1/check/puppet-openstack-unit-4.8-centos-7/a23fde8/ : SUCCESS in 13m 28s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/53/669653/1/check/puppet-openstack-unit-5.5-centos-7/5059bb5/ : SUCCESS in 12m 08s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/53/669653/1/check/puppet-openstack-beaker-centos-7/c169927/ : SUCCESS in 23m 36s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-beaker-ubuntu-bionic/021cf2b/ : FAILURE in 7m 49s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/53/669653/1/check/build-openstack-releasenotes/8bd2593/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"973639180e2c31c348b00daf0ca1ede2ee285683","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 09:21:46.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit b9e77891468299f8028417057070df5d1780794d","accounts_in_message":[],"_revision_number":1},{"id":"44cd89d968bcd20fc80b0e23ecf2764ce0914105","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 11:20:42.000000000","message":"Topic changed from OctaviaServerCertsKeyPassphrase-32chars-stable/stein to OctaviaServerCertsKeyPassphrase-32chars","accounts_in_message":[],"_revision_number":1},{"id":"143b1dc742a39eaa8e3cff12173c45531e3761ea","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 20:09:26.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"38c8d869074c17e8dcf925e18616d6eebb0730b7","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-10 07:56:32.000000000","message":"Abandoned\n\nTemporarily abandoning this patch to merge https://review.opendev.org/#/c/666987/ into master. will restore afterwards.","accounts_in_message":[],"_revision_number":1},{"id":"1635b158a9dbba3a10877f62be2a7fba2ef008de","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-10 13:21:23.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"abb07708d45d584a89beb1753138406aede3e6ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 13:49:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-lint-ubuntu-bionic/34e0134/ : SUCCESS in 6m 24s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-syntax-4-ubuntu-bionic/d939d16/ : SUCCESS in 6m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-syntax-5-ubuntu-bionic/91bcb7d/ : SUCCESS in 5m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-syntax-6-ubuntu-bionic/86fe4dc/ : SUCCESS in 4m 28s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/53/669653/1/check/puppet-openstack-unit-4.8-centos-7/7f62eef/ : SUCCESS in 12m 05s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/53/669653/1/check/puppet-openstack-unit-5.5-centos-7/45e88db/ : SUCCESS in 12m 03s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/53/669653/1/check/puppet-openstack-beaker-centos-7/861584c/ : SUCCESS in 21m 27s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/53/669653/1/check/puppet-openstack-beaker-ubuntu-bionic/5a0e898/ : FAILURE in 7m 58s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/53/669653/1/check/build-openstack-releasenotes/21091e9/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e5c7c46b8d719b41977980eca323f36a2482eb83","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-10 14:51:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"35cfda6be48d1e5a7b0d5b5b773ae2ad4f3bb837","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-10 17:17:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"852212e98d03c67515577ca442c3510557fc5399","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-10 17:40:37.000000000","message":"Patch Set 1:\n\nthis one needs W+1 please :-)","accounts_in_message":[],"_revision_number":1},{"id":"cf047f088bdb653d36fc26b60e11032961cb8567","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-10 17:56:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baf78b4a1da4f89f71541e3c28e32c983d0a8bff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:56:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ec5b5204922cc3c7612eb408e5828473334e6c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:29:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/53/669653/1/gate/puppet-openstack-lint-ubuntu-bionic/2762acd/ : SUCCESS in 4m 51s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/53/669653/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/519f5dd/ : SUCCESS in 3m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/53/669653/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/fa2239c/ : SUCCESS in 5m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/53/669653/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/f7ba1af/ : SUCCESS in 4m 21s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/53/669653/1/gate/puppet-openstack-unit-4.8-centos-7/e56ca29/ : SUCCESS in 29m 39s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/53/669653/1/gate/puppet-openstack-unit-5.5-centos-7/e785319/ : SUCCESS in 10m 55s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/53/669653/1/gate/puppet-openstack-beaker-centos-7/fab7e00/ : SUCCESS in 21m 43s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/53/669653/1/gate/puppet-openstack-beaker-ubuntu-bionic/183f182/ : FAILURE in 13m 59s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/53/669653/1/gate/build-openstack-releasenotes/d0c783c/html/ : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6f4544cf252542322edc0f9efa8749db14dc892c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:29:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"381f1a34daec7001b4a7825bc708a4314619806a","revisions":{"381f1a34daec7001b4a7825bc708a4314619806a":{"kind":"REWORK","_number":1,"created":"2019-07-08 12:58:40.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/53/669653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/53/669653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/669653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/669653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/669653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/53/669653/1"}}},"commit":{"parents":[{"commit":"59081ddd6cc514fed1b1c40ca53fbf1cb866e94a","subject":"Configure server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/59081ddd6cc514fed1b1c40ca53fbf1cb866e94a"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-06-23 13:39:36.000000000","tz":180},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-07-08 12:58:40.000000000","tz":0},"subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","message":"Configure 32 chars length server_certs_key_passphrase for Octavia\n\nThis change is related to I886f2b8ac7092d9b3da38852e92a615d5666eea7\nand Ibcdbe2605a7cabe3a5ef8245b4460c8f70220989.\n\nRelated-Bug: #1833942\n\nChange-Id: I5c2629d9e7700fe1dd6f915bc257b1f058e40617\n(cherry picked from commit d9564d7c235bc3073ef6eb444a6f1561fb6dffc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/381f1a34daec7001b4a7825bc708a4314619806a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/381f1a34daec7001b4a7825bc708a4314619806a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
