)]}'
{"id":"openstack%2Fpuppet-octavia~669822","triplet_id":"openstack%2Fpuppet-octavia~stable%2Frocky~I5c2629d9e7700fe1dd6f915bc257b1f058e40617","project":"openstack/puppet-octavia","branch":"stable/rocky","topic":"OctaviaServerCertsKeyPassphrase-32chars","hashtags":[],"change_id":"I5c2629d9e7700fe1dd6f915bc257b1f058e40617","subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","status":"MERGED","created":"2019-07-09 09:21:46.000000000","updated":"2019-07-29 06:45:25.000000000","submitted":"2019-07-29 06:45:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669822-1564382725209-652102b1","meta_rev_id":"117b52499045c2833ab0296ce5508db4ea04af87","_number":669822,"virtual_id_number":669822,"owner":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-29 06:45:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-07-29 00:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-07-29 05:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-07-28 16:58:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-07-28 08:42:11.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2019-07-28 16:58:11.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":1,"date":"2019-07-29 05:58:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-28 16:58:08.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-28 16:58:11.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-07-29 00:34:03.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-07-29 05:58:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-29 06:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f9528807cc8ce4a44d313f45ee6a4da3ba6f198","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 09:21:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"82d2754c34ac1cf70c9dc50f9a973ec52b60e474","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 09:21:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3ce54440586e44dfc504137fc62323f85310b682","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 09:23:06.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit ae4c7f458ac28fc28bccd072522d84349aca10a7","accounts_in_message":[],"_revision_number":1},{"id":"286802b2e99dbd61892fa63e9aca3c37c207c61c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 09:47:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/22/669822/1/check/puppet-openstack-lint/c97b9fe/ : SUCCESS in 4m 53s\n- puppet-openstack-syntax-4 http://logs.openstack.org/22/669822/1/check/puppet-openstack-syntax-4/83bab93/ : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5 http://logs.openstack.org/22/669822/1/check/puppet-openstack-syntax-5/2b3d9ee/ : SUCCESS in 5m 05s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/22/669822/1/check/puppet-openstack-unit-4.8-centos-7/9708ac2/ : SUCCESS in 13m 19s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/669822/1/check/puppet-openstack-beaker-centos-7/595e563/ : SUCCESS in 23m 04s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/669822/1/check/puppet-openstack-beaker-ubuntu-bionic/44e5725/ : SUCCESS in 14m 12s\n- build-openstack-releasenotes http://logs.openstack.org/22/669822/1/check/build-openstack-releasenotes/d8d894d/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a9ab8400179c2dae3c6e00b045a8f16eef30aa50","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 11:21:11.000000000","message":"Topic changed from OctaviaServerCertsKeyPassphrase-32chars-stable/stein-stable/rocky to OctaviaServerCertsKeyPassphrase-32chars","accounts_in_message":[],"_revision_number":1},{"id":"b68caa6143e4ff1af0f53f6d86ebec3303fd9439","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-10 07:56:37.000000000","message":"Abandoned\n\nTemporarily abandoning this patch to merge https://review.opendev.org/#/c/666987/ into master. will restore afterwards.","accounts_in_message":[],"_revision_number":1},{"id":"8afa6f3a47bec5541d24cd1293cd2bb7e2021b81","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-11 05:55:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"5b7eace6ad02fd1e6ee3579f0ee523ef353ada72","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-11 05:55:14.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"c55d9f952edae676bdd04d6e4508626978a0cbf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 06:23:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/22/669822/1/check/puppet-openstack-lint/3eb734f/ : SUCCESS in 6m 42s\n- puppet-openstack-syntax-4 http://logs.openstack.org/22/669822/1/check/puppet-openstack-syntax-4/2c3aa74/ : SUCCESS in 5m 59s\n- puppet-openstack-syntax-5 http://logs.openstack.org/22/669822/1/check/puppet-openstack-syntax-5/1511241/ : SUCCESS in 6m 05s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/22/669822/1/check/puppet-openstack-unit-4.8-centos-7/c3bb376/ : SUCCESS in 17m 49s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/669822/1/check/puppet-openstack-beaker-centos-7/e3e5b10/ : FAILURE in 20m 56s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/669822/1/check/puppet-openstack-beaker-ubuntu-bionic/d66fc26/ : SUCCESS in 18m 14s\n- build-openstack-releasenotes http://logs.openstack.org/22/669822/1/check/build-openstack-releasenotes/cee2d9b/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d52b7e35f2d031d51b9d39ec950b51921aa81a38","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-23 13:53:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"a7ec1d669d29401b795a049cb82b1f12bcaabdb9","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-28 08:34:29.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"ee70203a4047cf56165cf971707cad9bc8ed6c3a","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-28 08:34:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a3359d52f56d4387ca8ec2cd846560c5ee9b236d","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-28 08:37:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aac7abcc6f680e90662b9de84886f67388019913","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-28 08:42:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"71eface04ff92ad43079a1ecaaa08191c3ed43ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 09:18:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/22/669822/5/check/puppet-openstack-lint/a062b5a/ : SUCCESS in 6m 22s\n- puppet-openstack-syntax-4 http://logs.openstack.org/22/669822/5/check/puppet-openstack-syntax-4/4a4867e/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5 http://logs.openstack.org/22/669822/5/check/puppet-openstack-syntax-5/60689e5/ : SUCCESS in 6m 12s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/22/669822/5/check/puppet-openstack-unit-4.8-centos-7/3ac8db7/ : SUCCESS in 18m 14s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/669822/5/check/puppet-openstack-beaker-centos-7/b28a2a7/ : SUCCESS in 35m 12s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/669822/5/check/puppet-openstack-beaker-ubuntu-bionic/56cb3f7/ : SUCCESS in 17m 14s\n- build-openstack-releasenotes http://logs.openstack.org/22/669822/5/check/build-openstack-releasenotes/9a80324/html/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":5},{"id":"4717e66f76530dfb30754e33fd966969e2e4ab94","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-29 00:34:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f3a7937fbaeec875fdc2ac75cdc051e966ac2760","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-29 05:58:37.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3fb96632361c335ac15972140a841dd7656b89b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 05:58:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"54b6b09d7930e1f8f7c66daa8fe716bdf5d72b12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 06:45:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/22/669822/5/gate/puppet-openstack-lint/07bce74/ : SUCCESS in 5m 10s\n- puppet-openstack-syntax-4 http://logs.openstack.org/22/669822/5/gate/puppet-openstack-syntax-4/6b026e7/ : SUCCESS in 6m 42s\n- puppet-openstack-syntax-5 http://logs.openstack.org/22/669822/5/gate/puppet-openstack-syntax-5/a17349e/ : SUCCESS in 5m 44s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/22/669822/5/gate/puppet-openstack-unit-4.8-centos-7/abb7148/ : SUCCESS in 16m 03s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/669822/5/gate/puppet-openstack-beaker-centos-7/0101b87/ : SUCCESS in 26m 17s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/669822/5/gate/puppet-openstack-beaker-ubuntu-bionic/a0083d5/ : SUCCESS in 17m 23s\n- build-openstack-releasenotes http://logs.openstack.org/22/669822/5/gate/build-openstack-releasenotes/e808cdf/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":5},{"id":"117b52499045c2833ab0296ce5508db4ea04af87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 06:45:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"39b9d44e47309a485fdcdd35b5862159d58f9f1d","revisions":{"b9e77891468299f8028417057070df5d1780794d":{"kind":"REWORK","_number":1,"created":"2019-07-09 09:21:46.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/22/669822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/669822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/1"}}},"commit":{"parents":[{"commit":"733344b0786134fc9e083570cc366e6281f9f248","subject":"Configure server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/733344b0786134fc9e083570cc366e6281f9f248"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-06-23 13:39:36.000000000","tz":180},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-07-09 09:21:46.000000000","tz":0},"subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","message":"Configure 32 chars length server_certs_key_passphrase for Octavia\n\nThis change is related to I886f2b8ac7092d9b3da38852e92a615d5666eea7\nand Ibcdbe2605a7cabe3a5ef8245b4460c8f70220989.\n\nRelated-Bug: #1833942\n\nChange-Id: I5c2629d9e7700fe1dd6f915bc257b1f058e40617\n(cherry picked from commit d9564d7c235bc3073ef6eb444a6f1561fb6dffc5)\n(cherry picked from commit 381f1a34daec7001b4a7825bc708a4314619806a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b9e77891468299f8028417057070df5d1780794d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b9e77891468299f8028417057070df5d1780794d"}]},"branch":"refs/heads/stable/rocky"},"f3eb9cf8222a0f51ad0fcdde5dce2824664b779b":{"kind":"NO_CHANGE","_number":3,"created":"2019-07-28 08:34:37.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/22/669822/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/669822/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/3"}}},"commit":{"parents":[{"commit":"733344b0786134fc9e083570cc366e6281f9f248","subject":"Configure server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/733344b0786134fc9e083570cc366e6281f9f248"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-06-23 13:39:36.000000000","tz":180},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-07-28 08:34:37.000000000","tz":0},"subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","message":"Configure 32 chars length server_certs_key_passphrase for Octavia\n\nThis change is related to I886f2b8ac7092d9b3da38852e92a615d5666eea7\nand Ibcdbe2605a7cabe3a5ef8245b4460c8f70220989.\n\nRelated-Bug: #1833942\n\nChange-Id: I5c2629d9e7700fe1dd6f915bc257b1f058e40617\n(cherry picked from commit d9564d7c235bc3073ef6eb444a6f1561fb6dffc5)\n(cherry picked from commit 381f1a34daec7001b4a7825bc708a4314619806a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f3eb9cf8222a0f51ad0fcdde5dce2824664b779b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f3eb9cf8222a0f51ad0fcdde5dce2824664b779b"}]},"branch":"refs/heads/stable/rocky"},"769299ac2412d144025bcee91a58e746ceb67198":{"kind":"NO_CHANGE","_number":4,"created":"2019-07-28 08:37:17.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/22/669822/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/669822/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/4"}}},"commit":{"parents":[{"commit":"733344b0786134fc9e083570cc366e6281f9f248","subject":"Configure server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/733344b0786134fc9e083570cc366e6281f9f248"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-06-23 13:39:36.000000000","tz":180},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-07-28 08:37:17.000000000","tz":0},"subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","message":"Configure 32 chars length server_certs_key_passphrase for Octavia\n\nThis change is related to I886f2b8ac7092d9b3da38852e92a615d5666eea7\nand Ibcdbe2605a7cabe3a5ef8245b4460c8f70220989.\n\nRelated-Bug: #1833942\n\nChange-Id: I5c2629d9e7700fe1dd6f915bc257b1f058e40617\n(cherry picked from commit d9564d7c235bc3073ef6eb444a6f1561fb6dffc5)\n(cherry picked from commit 381f1a34daec7001b4a7825bc708a4314619806a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/769299ac2412d144025bcee91a58e746ceb67198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/769299ac2412d144025bcee91a58e746ceb67198"}]},"branch":"refs/heads/stable/rocky"},"39b9d44e47309a485fdcdd35b5862159d58f9f1d":{"kind":"NO_CHANGE","_number":5,"created":"2019-07-28 08:42:11.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/22/669822/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/669822/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/669822/5"}}},"commit":{"parents":[{"commit":"733344b0786134fc9e083570cc366e6281f9f248","subject":"Configure server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/733344b0786134fc9e083570cc366e6281f9f248"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-06-23 13:39:36.000000000","tz":180},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-07-28 08:42:11.000000000","tz":0},"subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","message":"Configure 32 chars length server_certs_key_passphrase for Octavia\n\nThis change is related to I886f2b8ac7092d9b3da38852e92a615d5666eea7\nand Ibcdbe2605a7cabe3a5ef8245b4460c8f70220989.\n\nRelated-Bug: #1833942\n\nChange-Id: I5c2629d9e7700fe1dd6f915bc257b1f058e40617\n(cherry picked from commit d9564d7c235bc3073ef6eb444a6f1561fb6dffc5)\n(cherry picked from commit 381f1a34daec7001b4a7825bc708a4314619806a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39b9d44e47309a485fdcdd35b5862159d58f9f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39b9d44e47309a485fdcdd35b5862159d58f9f1d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
