)]}'
{"id":"openstack%2Fpuppet-octavia~669824","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fqueens~I5c2629d9e7700fe1dd6f915bc257b1f058e40617","project":"openstack/puppet-octavia","branch":"stable/queens","topic":"OctaviaServerCertsKeyPassphrase-32chars","hashtags":[],"change_id":"I5c2629d9e7700fe1dd6f915bc257b1f058e40617","subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","status":"MERGED","created":"2019-07-09 09:23:06.000000000","updated":"2019-08-01 01:13:44.000000000","submitted":"2019-08-01 01:13:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669824-1564622024538-cc82055b","meta_rev_id":"849b86aabdfac113d8bd207fd1d899bbbe22477d","_number":669824,"virtual_id_number":669824,"owner":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-31 12:28:01.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-01 01:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-08-01 00:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-31 20:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":1,"date":"2019-08-01 00:40:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 20:08:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-01 00:40:16.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-08-01 01:13:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"647371a5f7f3a14cafca877ce4cfc73a9dabdbad","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 09:23:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"2f3553e6fa00d9d97425bfe0025c929a7745b68b","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 09:23:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0222306391af1ac43a3aa8a0eb82c0ef1acf6c11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 10:11:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/24/669824/1/check/puppet-openstack-lint/d868392/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4 http://logs.openstack.org/24/669824/1/check/puppet-openstack-syntax-4/32f14f4/ : SUCCESS in 4m 18s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/24/669824/1/check/puppet-openstack-unit-4.8-centos-7/afb6530/ : SUCCESS in 15m 11s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/24/669824/1/check/puppet-openstack-unit-5.5-centos-7/387de74/ : SUCCESS in 13m 36s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/24/669824/1/check/puppet-openstack-beaker-centos-7/7f58672/ : SUCCESS in 40m 11s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/24/669824/1/check/puppet-openstack-beaker-xenial/c3b3132/ : SUCCESS in 18m 34s\n- build-openstack-releasenotes http://logs.openstack.org/24/669824/1/check/build-openstack-releasenotes/c5d623f/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"087f322ceb095a4e907bbf19d39e118f47d00273","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-09 11:21:16.000000000","message":"Topic changed from OctaviaServerCertsKeyPassphrase-32chars-stable/stein-stable/rocky-stable/queens to OctaviaServerCertsKeyPassphrase-32chars","accounts_in_message":[],"_revision_number":1},{"id":"4614a633143e2357df4a1b3ae4857cf3f8709b07","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-10 07:56:54.000000000","message":"Abandoned\n\nTemporarily abandoning this patch to merge https://review.opendev.org/#/c/666987/ into master. will restore afterwards.","accounts_in_message":[],"_revision_number":1},{"id":"61ad16de055bd6c14967e9bbb59a072c15f1ae7e","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-29 11:50:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"98b7ebc4c40e29c3017073863c7c6bfb918ab6a1","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-29 11:50:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"169ede7a46d8a893edb26130217ae20ed4d8573d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 12:22:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/24/669824/2/check/puppet-openstack-lint/727acde/ : SUCCESS in 4m 58s\n- puppet-openstack-syntax-4 http://logs.openstack.org/24/669824/2/check/puppet-openstack-syntax-4/1059c6a/ : SUCCESS in 4m 51s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/24/669824/2/check/puppet-openstack-unit-4.8-centos-7/5a4cbef/ : SUCCESS in 18m 14s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/24/669824/2/check/puppet-openstack-unit-5.5-centos-7/4d58c4c/ : FAILURE in 10m 03s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/24/669824/2/check/puppet-openstack-beaker-centos-7/56044aa/ : FAILURE in 10m 24s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/24/669824/2/check/puppet-openstack-beaker-xenial/fa11485/ : SUCCESS in 16m 17s\n- build-openstack-releasenotes http://logs.openstack.org/24/669824/2/check/build-openstack-releasenotes/6881ec3/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"81ba8bf60891dae5b219a206fbfa0c2267a3986a","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-31 12:28:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f2e720b48e9ef171f498be55486d157dc5b138b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 13:06:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/24/669824/2/check/puppet-openstack-lint/fe315ba/ : SUCCESS in 5m 01s\n- puppet-openstack-syntax-4 http://logs.openstack.org/24/669824/2/check/puppet-openstack-syntax-4/8b02d98/ : SUCCESS in 6m 18s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/24/669824/2/check/puppet-openstack-unit-4.8-centos-7/bdbae03/ : SUCCESS in 18m 46s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/24/669824/2/check/puppet-openstack-unit-5.5-centos-7/8042bb8/ : SUCCESS in 13m 44s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/24/669824/2/check/puppet-openstack-beaker-centos-7/6d5732a/ : SUCCESS in 20m 53s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/24/669824/2/check/puppet-openstack-beaker-xenial/2e2f6a1/ : SUCCESS in 13m 30s\n- build-openstack-releasenotes http://logs.openstack.org/24/669824/2/check/build-openstack-releasenotes/da5efa4/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"447b3363a43235f8d816b874e7963651d9311a2f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-31 20:08:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1b299638f0887fec04897b9c0355784f3d82a1c9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-01 00:40:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"327e1d43fe713d1189f2db219b663b3cefffd393","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 00:40:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14ea3abec14bd1d9379c351b866c88149086cfc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:13:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"849b86aabdfac113d8bd207fd1d899bbbe22477d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:13:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://logs.opendev.org/24/669824/2/gate/puppet-openstack-lint/78eca62/ : SUCCESS in 5m 57s\n- puppet-openstack-syntax-4 https://logs.opendev.org/24/669824/2/gate/puppet-openstack-syntax-4/38ab03d/ : SUCCESS in 5m 03s\n- puppet-openstack-unit-4.8-centos-7 https://logs.opendev.org/24/669824/2/gate/puppet-openstack-unit-4.8-centos-7/de34923/ : SUCCESS in 16m 18s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/24/669824/2/gate/puppet-openstack-unit-5.5-centos-7/d3608bc/ : SUCCESS in 12m 00s\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/24/669824/2/gate/puppet-openstack-beaker-centos-7/6af64d1/ : SUCCESS in 25m 46s\n- puppet-openstack-beaker-xenial https://logs.opendev.org/24/669824/2/gate/puppet-openstack-beaker-xenial/3d60b1b/ : SUCCESS in 15m 07s\n- build-openstack-releasenotes https://logs.opendev.org/24/669824/2/gate/build-openstack-releasenotes/6b3b950/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4967222fc785f899e80c4c96b50d47d26625db35","revisions":{"ae4c7f458ac28fc28bccd072522d84349aca10a7":{"kind":"REWORK","_number":1,"created":"2019-07-09 09:23:06.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/24/669824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/24/669824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/1"}}},"commit":{"parents":[{"commit":"bdcd8132a1ef4e10b4c21832ca95fe890ea6badd","subject":"Configure server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bdcd8132a1ef4e10b4c21832ca95fe890ea6badd"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-06-23 13:39:36.000000000","tz":180},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-07-09 09:23:06.000000000","tz":0},"subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","message":"Configure 32 chars length server_certs_key_passphrase for Octavia\n\nThis change is related to I886f2b8ac7092d9b3da38852e92a615d5666eea7\nand Ibcdbe2605a7cabe3a5ef8245b4460c8f70220989.\n\nRelated-Bug: #1833942\n\nChange-Id: I5c2629d9e7700fe1dd6f915bc257b1f058e40617\n(cherry picked from commit d9564d7c235bc3073ef6eb444a6f1561fb6dffc5)\n(cherry picked from commit 381f1a34daec7001b4a7825bc708a4314619806a)\n(cherry picked from commit b9e77891468299f8028417057070df5d1780794d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ae4c7f458ac28fc28bccd072522d84349aca10a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ae4c7f458ac28fc28bccd072522d84349aca10a7"}]},"branch":"refs/heads/stable/queens"},"4967222fc785f899e80c4c96b50d47d26625db35":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-29 11:50:25.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/24/669824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/24/669824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/24/669824/2"}}},"commit":{"parents":[{"commit":"bdcd8132a1ef4e10b4c21832ca95fe890ea6badd","subject":"Configure server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bdcd8132a1ef4e10b4c21832ca95fe890ea6badd"}]}],"author":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-06-23 13:39:36.000000000","tz":180},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-07-29 11:50:25.000000000","tz":0},"subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","message":"Configure 32 chars length server_certs_key_passphrase for Octavia\n\nThis change is related to I886f2b8ac7092d9b3da38852e92a615d5666eea7\nand Ibcdbe2605a7cabe3a5ef8245b4460c8f70220989.\n\nRelated-Bug: #1833942\n\nChange-Id: I5c2629d9e7700fe1dd6f915bc257b1f058e40617\n(cherry picked from commit d9564d7c235bc3073ef6eb444a6f1561fb6dffc5)\n(cherry picked from commit 381f1a34daec7001b4a7825bc708a4314619806a)\n(cherry picked from commit 39b9d44e47309a485fdcdd35b5862159d58f9f1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4967222fc785f899e80c4c96b50d47d26625db35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4967222fc785f899e80c4c96b50d47d26625db35"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
