)]}'
{"id":"openstack%2Fpuppet-octavia~673512","triplet_id":"openstack%2Fpuppet-octavia~master~I8be48a7bd07e0a00972b4bef268c4bd2badc6711","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1838410","hashtags":[],"change_id":"I8be48a7bd07e0a00972b4bef268c4bd2badc6711","subject":"Add support for connection retry variables","status":"MERGED","created":"2019-07-30 13:05:01.000000000","updated":"2019-09-21 10:37:01.000000000","submitted":"2019-08-05 01:08:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673512-1564967316024-c33df9d0","meta_rev_id":"b199d03ee1c31011ead449d1df5e486b39023576","_number":673512,"virtual_id_number":673512,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-05 01:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-02 15:48:44.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-02 13:12:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-05 00:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-08-05 00:44:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 13:12:25.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-02 15:48:44.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-05 00:44:32.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-08-05 01:08:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84a8743102ef07633f8f1353cfa7ed27705c294d","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-30 13:05:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c74fb3a722b31d8552951df4e7766dbd07651d67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 13:42:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/12/673512/1/check/puppet-openstack-lint-ubuntu-bionic/dacce42/ : SUCCESS in 4m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/12/673512/1/check/puppet-openstack-syntax-5-ubuntu-bionic/e81e4cd/ : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/12/673512/1/check/puppet-openstack-syntax-6-ubuntu-bionic/0338d36/ : SUCCESS in 5m 06s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/12/673512/1/check/puppet-openstack-unit-5.5-centos-7/ec35520/ : FAILURE in 21m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/12/673512/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/25094d2/ : FAILURE in 13m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/12/673512/1/check/puppet-openstack-unit-latest-ubuntu-bionic/e6b61f5/ : FAILURE in 11m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/12/673512/1/check/puppet-openstack-beaker-centos-7/d4260ef/ : SUCCESS in 23m 08s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/12/673512/1/check/puppet-openstack-beaker-ubuntu-bionic/a878e5a/ : FAILURE in 13m 07s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/12/673512/1/check/build-openstack-releasenotes/d217bb1/html/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6b48acd75207f57945030dff883dd9da6c721341","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-02 12:21:56.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"01ef2243df463d1324ae5505eccf31f0c77d12eb","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-02 12:44:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17c9dde6e849ff5292a1ebb83dd15dfde66dd90f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 13:12:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/12/673512/2/check/puppet-openstack-lint-ubuntu-bionic/c802691/ : SUCCESS in 5m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/12/673512/2/check/puppet-openstack-syntax-5-ubuntu-bionic/86c14e5/ : SUCCESS in 5m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/12/673512/2/check/puppet-openstack-syntax-6-ubuntu-bionic/9638fe2/ : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/12/673512/2/check/puppet-openstack-unit-5.5-centos-7/ff9c463/ : SUCCESS in 20m 35s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/12/673512/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/16d1e2a/ : SUCCESS in 13m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/12/673512/2/check/puppet-openstack-unit-latest-ubuntu-bionic/027e722/ : SUCCESS in 10m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/12/673512/2/check/puppet-openstack-beaker-centos-7/afd1ee9/ : SUCCESS in 20m 39s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/12/673512/2/check/puppet-openstack-beaker-ubuntu-bionic/4f3a3cb/ : FAILURE in 8m 27s (non-voting)\n- build-openstack-releasenotes https://logs.opendev.org/12/673512/2/check/build-openstack-releasenotes/8ee6610/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"fc8f96c02b5b0ad2a0c11451fe7dd87809638a6c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-02 13:12:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a533cc3c7076f6dad87a127ad2430b64116fb71c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-02 15:48:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"efc4b397ec71e0f252f93a68d749c12691eb96e1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-05 00:44:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df17420565ed4d17bf3a61f1302a45c7fc0daa86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 00:44:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49bb8e6a0077f907c6e5a1180b18c3511f785b5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 01:08:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/12/673512/2/gate/puppet-openstack-lint-ubuntu-bionic/90cfbcd/ : SUCCESS in 4m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/12/673512/2/gate/puppet-openstack-syntax-5-ubuntu-bionic/629e6a7/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/12/673512/2/gate/puppet-openstack-syntax-6-ubuntu-bionic/c464459/ : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/12/673512/2/gate/puppet-openstack-unit-5.5-centos-7/43067e0/ : SUCCESS in 12m 27s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/12/673512/2/gate/puppet-openstack-unit-6.4-ubuntu-bionic/1deaaca/ : SUCCESS in 11m 27s\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/12/673512/2/gate/puppet-openstack-beaker-centos-7/32b83bf/ : SUCCESS in 20m 48s\n- build-openstack-releasenotes https://logs.opendev.org/12/673512/2/gate/build-openstack-releasenotes/bbea833/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"5c3e009e7b7fdfc1adeef36f36b85e054efda43b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 01:08:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ce126317367bbb6683bcf6673e678d29f2d5ada5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-21 10:37:01.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 8652716babb20258235544f3fd0591194fc5c0a4","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36ca9c36cbd414c056d87835a384e821f7b41fcd","revisions":{"a78ba87bfd162d2d52420fda530d23f5bb17886d":{"kind":"REWORK","_number":1,"created":"2019-07-30 13:05:01.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/12/673512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/12/673512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/1"}}},"commit":{"parents":[{"commit":"9e62de27cd39356a4fa17c815b92a5e6a2215cab","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9e62de27cd39356a4fa17c815b92a5e6a2215cab"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-30 13:04:02.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-30 13:05:00.000000000","tz":-240},"subject":"Add support for connection retry variables","message":"Add support for connection retry variables\n\nAdds support for connection_max_retries and connection_retry_interval\nconfiguration.\n\nChange-Id: I8be48a7bd07e0a00972b4bef268c4bd2badc6711\nCloses-Bug: #1838410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a78ba87bfd162d2d52420fda530d23f5bb17886d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a78ba87bfd162d2d52420fda530d23f5bb17886d"}]},"branch":"refs/heads/master"},"36ca9c36cbd414c056d87835a384e821f7b41fcd":{"kind":"REWORK","_number":2,"created":"2019-08-02 12:44:57.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/12/673512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/12/673512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/12/673512/2"}}},"commit":{"parents":[{"commit":"9e62de27cd39356a4fa17c815b92a5e6a2215cab","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9e62de27cd39356a4fa17c815b92a5e6a2215cab"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-30 13:04:02.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-08-02 12:44:20.000000000","tz":-240},"subject":"Add support for connection retry variables","message":"Add support for connection retry variables\n\nAdds support for connection_max_retries and connection_retry_interval\nconfiguration.\n\nChange-Id: I8be48a7bd07e0a00972b4bef268c4bd2badc6711\nCloses-Bug: #1838410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/36ca9c36cbd414c056d87835a384e821f7b41fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/36ca9c36cbd414c056d87835a384e821f7b41fcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
