)]}'
{"id":"openstack%2Fpuppet-octavia~676298","triplet_id":"openstack%2Fpuppet-octavia~master~Ie017fc4fa62f4ed56b1b643cc6fa4875acec0fe6","project":"openstack/puppet-octavia","branch":"master","topic":"dev","hashtags":[],"change_id":"Ie017fc4fa62f4ed56b1b643cc6fa4875acec0fe6","subject":"Add bugs url link to README","status":"ABANDONED","created":"2019-08-14 01:39:35.000000000","updated":"2019-08-15 08:28:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"625df7d8300b165914f39719322ff2314c8f75e3","_number":676298,"virtual_id_number":676298,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-08-14 01:48:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":-1,"date":"2019-08-15 08:05:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 01:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 08:05:19.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"49f6d340df5ad22bd475afd1a1512a9076303a05","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-14 01:39:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70bb9b02672fc6795b879db00cbe2b177ffc8b74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 01:48:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c61a251619c4c38b81218182c908ce6 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cafadd8617202b8623b39a307a36084685dc5da9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 08:05:19.000000000","message":"Patch Set 1: Code-Review-1\n\nWe shouldn\u0027t do this, we are planning to move to StoryBoard which would require us to change these all over again.","accounts_in_message":[],"_revision_number":1},{"id":"0b2ca88636d65c90af53b16977c0f7afeabbc32b","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-15 08:28:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1659517fc3b47641ddd37c85718a67ca1c35aaf","revisions":{"a1659517fc3b47641ddd37c85718a67ca1c35aaf":{"kind":"REWORK","_number":1,"created":"2019-08-14 01:39:35.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/98/676298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/98/676298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/98/676298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/98/676298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/98/676298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/98/676298/1"}}},"commit":{"parents":[{"commit":"211d8a71e4d289deaaf54139357af17dae918df3","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/211d8a71e4d289deaaf54139357af17dae918df3"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-14 01:37:36.000000000","tz":-420},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-14 01:39:31.000000000","tz":-420},"subject":"Add bugs url link to README","message":"Add bugs url link to README\n\nChange-Id: Ie017fc4fa62f4ed56b1b643cc6fa4875acec0fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a1659517fc3b47641ddd37c85718a67ca1c35aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a1659517fc3b47641ddd37c85718a67ca1c35aaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
