)]}'
{"id":"openstack%2Fpuppet-octavia~676553","triplet_id":"openstack%2Fpuppet-octavia~master~Iaf4ffdc4fb3eed847f78831230f2cdfa30f8d8d6","project":"openstack/puppet-octavia","branch":"master","topic":"remove_deprecate_pki_related_options","hashtags":[],"change_id":"Iaf4ffdc4fb3eed847f78831230f2cdfa30f8d8d6","subject":"Remove deprecated pki related options","status":"MERGED","created":"2019-08-15 03:52:33.000000000","updated":"2019-08-20 08:33:03.000000000","submitted":"2019-08-20 08:33:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676553-1566289983982-0e67f363","meta_rev_id":"f40c64b95032a612bc7bfc8778e2fefd1916a136","_number":676553,"virtual_id_number":676553,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-20 08:33:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-19 13:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-20 07:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-20 07:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 16:24:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-20 07:58:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-20 08:33:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2dc0f20f2ff701720230cc6eb68eff7c5c664a2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 03:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac51b35539f4e5aae8e34c97e773dc21f619ac22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 05:30:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/522d409f69d644f8ae7b7fc297bd3cc9 : SUCCESS in 5m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9513c9221bab4e55bb9c361c586550bb : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49d2c0482aec40dbb3fa588d7e619974 : SUCCESS in 4m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d319611fa059483492db3e2e8f6e9307 : SUCCESS in 18m 35s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f980b2df42bc4688a56e71fa65591776 : SUCCESS in 14m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8fb0e7c7ff941f69f6c0a0fe86ea7da : SUCCESS in 12m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/39a62e42bd20489ebb40323ed5a8f890 : TIMED_OUT in 1h 30m 48s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c1d871c81ec41498fc1c21938e95514 : FAILURE in 7m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69570faf424242af9aaa0d620b94530c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f2cd3c092bd92213a9359a8716246927ffecd79b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 05:36:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c455684c24c6ca3b9a6e637409c934312ced7c61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 07:12:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60c27772fec34ed6ae54c857d83f8c66 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7052490886064e49ac71a12952f63853 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce98563ad0be456da5e80c5f16766ee5 : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/318c6c4f7ada49afb801b8ff6b05fdcb : SUCCESS in 17m 43s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/758948781d5f4646a96f306ae025cf79 : SUCCESS in 11m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada5393897d544ae98310305faf0bdab : SUCCESS in 11m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/21692dc6bdea4230a7d14e61a0b66fde : TIMED_OUT in 1h 30m 45s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9872fd1ae8fe446db7d36b30cc378ace : FAILURE in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afa7f58876104895b37b9ba60b4c0000 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"da872a0431b9a580cb786c62ffdcb87a186ac251","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 16:24:56.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d3405d337099591788117173b7ad078f93fec496","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 18:27:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b47dad11a314267a098fdd66d58a908 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/287a67d3ecfa49309d00e96d07187c22 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fc1b791b9b6463bb1aba08c4971fd86 : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d1b03547471e4498a61c0faa71d1af9d : SUCCESS in 16m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8161c228b09e48eea455fd8d07e0ca32 : SUCCESS in 11m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff83ea1ddde74b8bb8120893886aeb4c : SUCCESS in 9m 19s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1d7aef961bb046c09f2160bf7cc3a5c7 : TIMED_OUT in 1h 31m 20s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92f8c85df8ea4bbe8dc4ea578ad5311a : FAILURE in 7m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d92ba016ef2442a6b9efe7faeb760f73 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d9ceb2129d4b8b0218849dfdad551d7caf3b91ac","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-16 05:57:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6cdff61d1122a04e2afe9ceb3f4555478c05bd77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 07:31:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb22a9ad5b90496ca826c8aa8edcffbb : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f38c7808f2194bbc94f2c3fc428cbdb8 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32dc02a965dc40ee8e22383669893b23 : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/98474d2a5d5841b8b4be6087f0e0d28b : SUCCESS in 14m 09s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4f6366be01844889a3f0c68324efec2 : SUCCESS in 10m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da7d172a21ed42c08c37f8875e74954c : SUCCESS in 9m 30s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4ca4a8d97fd94be18376ad7e9fd9dff3 : TIMED_OUT in 1h 31m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6a26b41e3f84b1b8119fa5c91193629 : FAILURE in 8m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf4cb7155e5743c98501ba42e302264f : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e7d60cd835415387eb30ea0bbe53de58fbdde60e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-16 08:16:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83447294e7934011d97e6812dd4156acf389938c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:50:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef60ee042c374ed59ec3e3912c7548ef : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/690581ffae634da9a58e46e009c14f10 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22b720fc6159483685f3b72d150cfd51 : SUCCESS in 4m 40s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2900a3fd64bd46caa1a4ffecf78cd44c : SUCCESS in 19m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94961ca0306a4b999cd312dc3f8fdb3f : SUCCESS in 10m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2baf8e3ebade4d2fbe07499e6dc9066a : SUCCESS in 14m 55s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4eae48b610414b3694e69f3a27b5151e : TIMED_OUT in 1h 31m 12s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d7219704f704698ac8d853aa4eff6c7 : FAILURE in 11m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f04eb6660b24dd3a8efc56ce92ca8e8 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"36b2dafdea1697511e89a27d4ae68ad7c2d8bf68","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-17 01:22:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af28d804f56141e4e8fcacda2369c3b23135323e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 02:56:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a3d8935354f4f3caa05bb8eee3c70ff : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37a138be66c642cb8280cee49c632ae8 : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7442cb000c6f4b2ab66e7dc90868a67f : SUCCESS in 5m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f4ea65bd437944b7894f01bd8184b961 : SUCCESS in 16m 05s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d42c2f798c6f471e918c7ed8446d4454 : SUCCESS in 12m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d61f47b2c54e486dba6aef93aaa4f3c9 : SUCCESS in 13m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/426761037c6348308fcbbbef1ccadf21 : TIMED_OUT in 1h 31m 20s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab2b85b76b8044dfadb60a01c50c09b8 : FAILURE in 12m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0ec401a499943879264e9fc363197a7 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"57ba37cc7eff00984a61f9563258bbf33f3188a5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-19 03:56:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f94b375373dcab5c2c56cd84e596db0827928bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 05:31:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3c9c480e65c4d36ae8042473cbc2438 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e20456e80a64303b3306d3da87b7c4a : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f710cc312324d90810f605f7f984df8 : SUCCESS in 4m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f28f2546472c4ea48456f18fe5951968 : SUCCESS in 13m 58s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2af5d78889a24a35be50c03ae784d42c : SUCCESS in 9m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b893f288992549019a0e2650ac714ffa : SUCCESS in 8m 44s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1166202fada744bd9692ed435f86b381 : TIMED_OUT in 1h 31m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69db83cf709e4854a8e9df18ee5e303a : FAILURE in 8m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab77ca068bc4b93a48b254825ff7b7e : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5a7422858d117f76dd21a62d41eacb26865e744c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-19 13:49:25.000000000","message":"Patch Set 1:\n\nlooks like we need the placement service for this too like puppet-nova. will fix","accounts_in_message":[],"_revision_number":1},{"id":"7d32c69f7362fee80e8138e8a8b73ca938adb1a6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-19 13:54:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2a10519d77177231c900b0177b30d15ed6acaefa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 14:30:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a17480f18f4644a588704f8cbb5a7466 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f720146b0b8549c48a658371d74ea010 : SUCCESS in 6m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5ab2a40f27742e09a614ed0ddfd70b5 : SUCCESS in 6m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2e7d2e13abab40668a4931c93a7c10ee : SUCCESS in 25m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1671c8e0b17e4fd182d8b0bbc2cdabfb : SUCCESS in 15m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d810f3993ff4462afb4a517b1b2b16f : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2661aed82f7b4bbc855a7282d1604020 : SUCCESS in 28m 50s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0698ce0bd814cbda4be7045d1b80185 : FAILURE in 8m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97946a6a7f194ef4ab8da30f63054e36 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8b22f1195ae10bcaa18ea736846bcca8fa73387e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-20 07:58:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"625a3338f7c18e3cc239c74ba23e1109857da789","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 07:59:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c96ab8e003d155e71ebec2ef10a742b17a1d4b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 08:33:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f94bb51e3eb64706e533d5694611b0a5ed68764d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 08:33:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5670cfce7d9c416a9de59ae9fe841b2e : SUCCESS in 5m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc77803d5ac341bfac40bd5960506663 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b53a87d75d9246e69b4039c931b2a2f0 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1e2c15bec2c945f59fa61596b7e841f4 : SUCCESS in 11m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7caf7e0a5a94de39987e24e41105988 : SUCCESS in 9m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/44f6c5a0aec54dce860495d3fc3eedb8 : SUCCESS in 31m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e378ed14f844cb19a00650e79875e98 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b098d97e3ad5fee9f30f7297190fd80bc4c1b966","revisions":{"792aed182888043f5221b30c442acd4733e620bf":{"kind":"REWORK","_number":1,"created":"2019-08-15 03:52:33.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/53/676553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/53/676553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/1"}}},"commit":{"parents":[{"commit":"211d8a71e4d289deaaf54139357af17dae918df3","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/211d8a71e4d289deaaf54139357af17dae918df3"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:38.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:38.000000000","tz":480},"subject":"Remove deprecated pki related options","message":"Remove deprecated pki related options\n\nThe deprecated pki related options check_revocations_for_cached and\nhash_algorithms option has been removed.\n\nChange-Id: Iaf4ffdc4fb3eed847f78831230f2cdfa30f8d8d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/792aed182888043f5221b30c442acd4733e620bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/792aed182888043f5221b30c442acd4733e620bf"}]},"branch":"refs/heads/master"},"b098d97e3ad5fee9f30f7297190fd80bc4c1b966":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-19 13:54:53.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/53/676553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/53/676553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/53/676553/2"}}},"commit":{"parents":[{"commit":"47e11f0999d147d5dac28760c11d02a514f41b24","subject":"Add placement service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/47e11f0999d147d5dac28760c11d02a514f41b24"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:38.000000000","tz":480},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-19 13:54:53.000000000","tz":0},"subject":"Remove deprecated pki related options","message":"Remove deprecated pki related options\n\nThe deprecated pki related options check_revocations_for_cached and\nhash_algorithms option has been removed.\n\nChange-Id: Iaf4ffdc4fb3eed847f78831230f2cdfa30f8d8d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b098d97e3ad5fee9f30f7297190fd80bc4c1b966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b098d97e3ad5fee9f30f7297190fd80bc4c1b966"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
