)]}'
{"id":"openstack%2Fpuppet-octavia~677750","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fstein~I1d72117dfa13c546e3b27ba67fb9e735d0f3b2f3","project":"openstack/puppet-octavia","branch":"stable/stein","topic":"fix_variable_name-stable/stein","hashtags":[],"change_id":"I1d72117dfa13c546e3b27ba67fb9e735d0f3b2f3","subject":"Deprecate spare_amphorae_pool_size","status":"MERGED","created":"2019-08-21 14:45:49.000000000","updated":"2019-08-22 20:27:04.000000000","submitted":"2019-08-22 20:27:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677750-1566505624254-b361f005","meta_rev_id":"e235e5869a6e2db6382a518e56cbfef06e17b80f","_number":677750,"virtual_id_number":677750,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-22 20:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 02:25:06.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-22 19:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-22 19:56:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-21 14:45:49.000000000","_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 14:45:49.000000000","updated_by":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"reviewer":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"state":"REVIEWER"},{"updated":"2019-08-22 02:25:06.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 19:56:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-22 20:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fad0d1ee26e69ae8be727d115b5280d6d4d9c32f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-21 14:45:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8fc403ccdcd079aaf6915b5083c3a71f17e930d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-21 14:45:57.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit e59bc8bef926334239b1a6e8d391fc080dfac3a0","accounts_in_message":[],"_revision_number":1},{"id":"d75be426a34328d0824206803b084780c9d2b819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:19:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1852d6c5db4d48cc8963f966afed716b : SUCCESS in 6m 54s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bda2ef50d064e999e601b222d87a6c4 : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf09df4735ff423ca34cc35009f40fbf : SUCCESS in 6m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/157606c1dfea43ac912f5af974a8d1c6 : SUCCESS in 7m 55s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/ff9e9d7200e1444988b6e6cc7d5fe399 : SUCCESS in 18m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c23dbbc3c7924187a5379ddcdd59e056 : SUCCESS in 15m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ca26b3d9877b41b093995663916f6d13 : SUCCESS in 27m 45s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9cb288d5ccc488583bde2c2a3ec2c14 : FAILURE in 10m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc1a628149ec45b5ab4f387c19486617 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0902e48b9d75e2bbfe5ed00598d9ce56c82bf657","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:25:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be839d03343528e3d1f9428ee2b02faab72c2d8a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 19:56:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f74bdac5bb050e7df443d686dac6fc72cf653426","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 19:56:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"364dcca5713ddc90b6bfa3c216665c534bed549f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:27:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1cd9c1294d8c9f27d92de910eb47d1e9d1166577","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:27:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6fc1544661e412d96a113f4ebc45dfa : SUCCESS in 5m 01s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdb52518ae6d4d1baa6121e1cbae6ac9 : SUCCESS in 6m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89381aab486a4399979451d9301c70ad : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc91b8b42e604f5790c680f910df08b9 : SUCCESS in 5m 00s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/128fb7dd97ca420aa83267675ccd7716 : SUCCESS in 21m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5750c1a7759540ec9fb74e57bfbde23f : SUCCESS in 19m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/882f04b9172a4c9096a414d7ae003fe4 : SUCCESS in 21m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92f9169933164b498208038afca11b31 : FAILURE in 8m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11011aabaaa5497d8bd20275a83b2a28 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e871346bf0b7642e8e0bf25d4de6977ce14cfe5c","revisions":{"e871346bf0b7642e8e0bf25d4de6977ce14cfe5c":{"kind":"REWORK","_number":1,"created":"2019-08-21 14:45:49.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/677750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/50/677750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/50/677750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/50/677750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/50/677750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/50/677750/1"}}},"commit":{"parents":[{"commit":"381f1a34daec7001b4a7825bc708a4314619806a","subject":"Configure 32 chars length server_certs_key_passphrase for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/381f1a34daec7001b4a7825bc708a4314619806a"}]}],"author":{"name":"Matthew J. Black","email":"mjblack@gmail.com","date":"2019-06-28 18:25:07.000000000","tz":-240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-21 14:45:49.000000000","tz":0},"subject":"Deprecate spare_amphorae_pool_size","message":"Deprecate spare_amphorae_pool_size\n\nIn octavia::housekeeping class the spare pool size variable\nis misspelled and has no effect.  A new parameter has been\nadded with the correct spelling. The configuration option\nin octavia conf is now correctly configured.\n\nChange-Id: I1d72117dfa13c546e3b27ba67fb9e735d0f3b2f3\n(cherry picked from commit 6a8720da4fc555b9fb4c7dd29ddedd753f3ec174)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e871346bf0b7642e8e0bf25d4de6977ce14cfe5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e871346bf0b7642e8e0bf25d4de6977ce14cfe5c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
