)]}'
{"id":"openstack%2Fpuppet-octavia~682636","triplet_id":"openstack%2Fpuppet-octavia~master~I5e2d2db8e7113e0c50b7cbd6a7ed383492477180","project":"openstack/puppet-octavia","branch":"master","hashtags":[],"change_id":"I5e2d2db8e7113e0c50b7cbd6a7ed383492477180","subject":"Add build_active_retries and port_detach_timeout","status":"MERGED","created":"2019-09-17 12:59:57.000000000","updated":"2019-09-19 14:21:41.000000000","submitted":"2019-09-19 14:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682636-1568902832327-4a385b24","meta_rev_id":"ecd28d34caf5688a6a6cbdf7b337ac3cf88903b2","_number":682636,"virtual_id_number":682636,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-09-19 14:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-18 00:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-17 15:44:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-19 13:47:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 15:44:51.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-09-18 00:37:01.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-09-19 13:47:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-19 14:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fa29635baf14615270ccbfffc27d9a178df60a3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-17 12:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cd0467ad5e86ff0d208a410ce253196e35ed959","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-17 13:02:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e73bff6f37aa9a8846f209f358d6ac6a6c7861df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:41:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5e45e5ab5a84db996f02767e9e75be1 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f2b9854b0604d298a038f8d420ee861 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16032897a48e4358861f6524b2dc294f : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d95db74c0c0548a38a9a49e0bdf16e88 : SUCCESS in 16m 44s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/508bfd4ed9c24f0593886281afaa94ce : SUCCESS in 17m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9da85ead053a4f5789c1e63951a39db5 : SUCCESS in 12m 03s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f226c1dc931549d4994f121a4252b30b : SUCCESS in 29m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d7945cb1cbc4b058101076be71bf450 : FAILURE in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1b526370553acf01c6ca4de6f1ddc2420568e5e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-17 15:35:02.000000000","message":"Patch Set 2: Code-Review+1\n\nnit: would be nice to have unit tests as well","accounts_in_message":[],"_revision_number":2},{"id":"0b6cbb3b732f85e24b2c1d71e25c8005c3c2eb17","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-17 15:42:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"69245a5c1030e90ebba130d5bd5d1778023c0f1c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-17 15:44:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"79e218f6d0d95be33705b4c3dbcf18ce0ef85127","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:14:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fe26b5fc62c4a749250d431b1c5a106 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f61608207ed4bedb68cedf8b2b53113 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13da636ad32f47bcbea8f99cac5372fc : SUCCESS in 5m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8556290dc49f45848c2a5d3aa28583ba : SUCCESS in 11m 54s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a1a5e1bb0f043118090cac4f02af2c0 : SUCCESS in 11m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba47346f09414cdd986a347baee2b9a9 : SUCCESS in 15m 40s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f52886e6f82940f39992fefd7e76782e : SUCCESS in 24m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96d5fd6261840bd8fd796009980306d : FAILURE in 7m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2a4302c088b4e1191665c3186f92a59 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"0fd21d21b7a51000aaaf304e57ab2503dcaab731","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-18 00:37:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"95a0731ca44a26d5faeb872f37267ceea0bad484","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-19 13:47:51.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4fb85a1557a65bd5c6066d7d3e7fde27402e79c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:50:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"75378b0fad13702a6a658f8973614cc580639ca7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:20:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2552c3dd5f6c42d4af4af945d9558aa7 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/146a43e61fe14bd284b9bad02d18a2b6 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9624d4283dcc48ff93d6f44b531bf33d : SUCCESS in 3m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/69e66344c9604e38ab9744da578e9d82 : SUCCESS in 10m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6594fa1c5d14848b8ffb414c1817171 : SUCCESS in 12m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/27833ac0e93348b4b0b3ad223d5ff5ae : SUCCESS in 24m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63f782856a964983a190b86e113cc2bc : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"80ee95c7aa5cee128dac42b4b43ae8993ef0a781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:20:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ecd28d34caf5688a6a6cbdf7b337ac3cf88903b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:21:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49881ef645fd4d9a97ce6ff23e05f115 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"96fa23758dfcd4cf7379bf1b229fd392c9c24bd7","revisions":{"eded6143de746eef55af19cf847798a715583107":{"kind":"REWORK","_number":1,"created":"2019-09-17 12:59:57.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/682636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/682636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/1"}}},"commit":{"parents":[{"commit":"31d89635ebcee7c7408509c375e72e72cde7c686","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/31d89635ebcee7c7408509c375e72e72cde7c686"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 12:57:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 12:57:50.000000000","tz":120},"subject":"Add parameter build_active_retries","message":"Add parameter build_active_retries\n\nParameter build_active_retries sets retry threshold for waiting for a\nbuild slot for an amphorae.\n\nChange-Id: I5e2d2db8e7113e0c50b7cbd6a7ed383492477180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/eded6143de746eef55af19cf847798a715583107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/eded6143de746eef55af19cf847798a715583107"}]},"branch":"refs/heads/master"},"e1b4bd0f43ed6954a1d47148057edf23f8f736aa":{"kind":"REWORK","_number":2,"created":"2019-09-17 13:02:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/682636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/682636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/2"}}},"commit":{"parents":[{"commit":"31d89635ebcee7c7408509c375e72e72cde7c686","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/31d89635ebcee7c7408509c375e72e72cde7c686"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 12:57:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 13:02:04.000000000","tz":120},"subject":"Add build_active_retries and port_detach_timeout","message":"Add build_active_retries and port_detach_timeout\n\nParameter build_active_retries sets retry threshold for waiting for a\nbuild slot for an amphorae.\n\nParameter port_detach_timeout sets seconds to wait for a port to detach\nfrom an amphora\n\nChange-Id: I5e2d2db8e7113e0c50b7cbd6a7ed383492477180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e1b4bd0f43ed6954a1d47148057edf23f8f736aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e1b4bd0f43ed6954a1d47148057edf23f8f736aa"}]},"branch":"refs/heads/master"},"96fa23758dfcd4cf7379bf1b229fd392c9c24bd7":{"kind":"REWORK","_number":3,"created":"2019-09-17 15:42:50.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/682636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/682636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/682636/3"}}},"commit":{"parents":[{"commit":"31d89635ebcee7c7408509c375e72e72cde7c686","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/31d89635ebcee7c7408509c375e72e72cde7c686"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 12:57:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 15:42:40.000000000","tz":120},"subject":"Add build_active_retries and port_detach_timeout","message":"Add build_active_retries and port_detach_timeout\n\nParameter build_active_retries sets retry threshold for waiting for a\nbuild slot for an amphorae.\n\nParameter port_detach_timeout sets seconds to wait for a port to detach\nfrom an amphora\n\nChange-Id: I5e2d2db8e7113e0c50b7cbd6a7ed383492477180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/96fa23758dfcd4cf7379bf1b229fd392c9c24bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/96fa23758dfcd4cf7379bf1b229fd392c9c24bd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
