)]}'
{"id":"openstack%2Fpuppet-octavia~683690","triplet_id":"openstack%2Fpuppet-octavia~master~I784c58d6898e843daa153aebb0c89482a7bb6050","project":"openstack/puppet-octavia","branch":"master","topic":"log-offload","hashtags":[],"change_id":"I784c58d6898e843daa153aebb0c89482a7bb6050","subject":"Add parameters for amphora log offloading","status":"MERGED","created":"2019-09-20 18:38:59.000000000","updated":"2019-10-10 15:07:33.000000000","submitted":"2019-10-10 15:06:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683690-1570719975501-94a06386","meta_rev_id":"1b7810a74727ae566069ba5eb79361944a309dd5","_number":683690,"virtual_id_number":683690,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 15:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-10-10 14:05:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-10-10 14:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-10-10 14:33:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-10 14:36:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 10:58:56.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-10 14:05:38.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-10 14:33:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-10 14:36:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-10 15:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f29b48b3af02cb74817c81258661b0a9e67ac94d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 18:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3966ed9954bfdd8f4e07a038b49391dca82da959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 19:10:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ce2d695f84a4e8980e2215388607acd : SUCCESS in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e27a7a46a8d04e569943eb53e1a8b648 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b1b4d3c77044759a2a193e40ea077f3 : SUCCESS in 3m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/23d35ea9043f41f7bc4f45c9c7691926 : SUCCESS in 12m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e211e7c91b4542e494b6e03ccdc36f31 : SUCCESS in 11m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a40690dbdee4d4db46dd369e09687b1 : SUCCESS in 10m 57s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3ba3c42ee2a444d68260ffa11252a278 : SUCCESS in 24m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b02ed0888aa24569b2a0bfed38ceec71 : FAILURE in 8m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e05e0df821b4b78b899e6885e0fe504 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"14ad90133387b8e576b33da4a5c973502656e6a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-20 19:17:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"60e5169eb3606f1e8443c1d9730a94f03e6b4c51","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 19:36:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4947c1b6eac042d1746873ab5db4675431cdcdf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 20:05:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/829f741cfaf943469ac4bbf486924084 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/369799723703464a9f303e01ae43066d : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/876c05304f4e49beab9b3c1a42bc79f2 : SUCCESS in 4m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3d7071de909a43dcbb05fdcc9940ddd4 : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daad98d0027c4ea19b46683902447457 : SUCCESS in 13m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63f11c3226bd43b1bfdfe68530cbffe9 : SUCCESS in 12m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/59f79fc4006848b08e637c6d52bcaef9 : SUCCESS in 24m 50s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7738cd7aa9a3411fbc7bdfde92f26c9b : FAILURE in 9m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca9cd2cda8904318b8318bd491f88f40 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4d14498a77865acc514a36e335fa36328f79c542","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-20 21:03:51.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"2731dfd75c570d5d4284475641084c1da88be95f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-23 10:58:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"af8ec3375966ac1c7c38eb92d138ee75da02a377","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-27 21:33:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1ffd3b575d25dd1f2c9c17057f50a17a132ac53a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-09 18:38:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9c0a8f33dc8a62e63512c60d27e600420396ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:55:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c6619182667484d925a7e76c92dc388 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a03d1b91ded54f3087062b95cb98ab58 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dd4ea17a09c40248b320fb5a2e13028 : SUCCESS in 3m 40s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ab5ca8d0dc41415092f3782aacb5a549 : FAILURE in 6m 34s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/936ba19121ec4221870ec481fce15d33 : SUCCESS in 10m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4f180d84cc4426bbf8ab2d59ede3c27 : SUCCESS in 11m 15s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/447d2f6591c944639857070045504152 : FAILURE in 6m 37s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42a637ca14c84e2b982ffebd86e41c9a : FAILURE in 6m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28fb9152412244b2a5e7b802562f6317 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"f630e71449abc1b8a806c4dd486c562a834b3606","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-10 10:08:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e088e99113919d5692a9fbcdc24af40e0f6356cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 10:36:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adbf6450678f429484cdc59f07ffb2b4 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29fe58fc92b54c0db6c8d174c276fe30 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1ea5aeff7dc4c0c94e03971d24d62c4 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/29a98d1a7d774d0c975c21ea9852142a : FAILURE in 8m 33s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/012d0e6cc00147c2a6e96bbd7bdeb40f : SUCCESS in 9m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe2c9e00f3274f93a8c3b51b7131bc74 : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f5a92061ea6f4b5da7a6d2696807d357 : SUCCESS in 22m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd28655f6e7846a3bd670227dfb545d5 : FAILURE in 7m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c38f7bc849d47ba90813cc537083bec : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"a83395904e72564bb5ca7e2e81140d3faf5dc312","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-10 10:46:08.000000000","message":"Patch Set 3:\n\nGem::InstallError: signet requires Ruby version \u003e\u003d 2.4.0.\n\nhttps://zuul.opendev.org/t/openstack/build/29a98d1a7d774d0c975c21ea9852142a/log/job-output.txt#2987\n\nI had expected https://review.opendev.org/#/c/687663/ to fix this but didn\u0027t?!","accounts_in_message":[],"_revision_number":3},{"id":"18716a9ea169b3c8a24a54774a8d1cec94e74a72","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-10 13:27:01.000000000","message":"Patch Set 3:\n\nrecheck\n\nshould be fixed now","accounts_in_message":[],"_revision_number":3},{"id":"92ba2138af90a68e6dd0203bdf085430e8d412dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:02:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b90c5ee5d2fc42e591ea9138014c7da0 : SUCCESS in 7m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26d50740e1664339994a5dee8bf2f2e4 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec47229dbae44b0bbb43fed03abe9fbb : SUCCESS in 5m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/22bac3292e2b463ca10fce09776f67f6 : SUCCESS in 17m 35s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74e021f91884407e9d4ff605a002fce6 : SUCCESS in 13m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bae8b3a824a4ff38365bb8588828403 : SUCCESS in 14m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0c05d3b54c8b4e3ab9a138a6bd0ab119 : SUCCESS in 32m 31s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9a8b8231d9f4d7ca2a9a350cc623ebc : FAILURE in 12m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f50b4efd86794c2fa1061c5ca41a2722 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":3},{"id":"3296ac196d9f74da0ae447ab891c9e9cf1b0ecc6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-10 14:05:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b4e34bb21c6a3d1eefdf96c76d785da27f4d17f5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-10 14:33:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"68ccc029739c64ac9678c681622311fbfbcdd113","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-10 14:36:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b516390579c13d95214c4e1ec01caec3f6c8e636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:36:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7c32e8ef6d9e315adc650c093bce6c402b2d3a15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:06:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e51e4b96215447deba5739844d7a8359 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71f47c3f4e294b2b86be398e3c19d605 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85b239cf8bbc4dc8bce3bd511f7ddb88 : SUCCESS in 6m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c7d4cc0752c8455ebdd652ec8c115e1f : SUCCESS in 19m 19s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbbd4e939adf40ab8253f323496e99e8 : SUCCESS in 12m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3e056e0406d347ac9365bce4905f1a79 : SUCCESS in 26m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8866f70e51bc40658d7a1838c0270e6c : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"9bcad0577357e5600e0eba31579f858e1b63b3f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:06:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f7b950df85ca7d0e7d897398627f43b88f164bba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:07:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3a28e4bda9f4e069664ad692a656e2d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"952e448d8fb141be1ccd50eb12ca05cc80a78f15","revisions":{"726893199449ba830dc8904f6d9b95a8161de79b":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:38:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/683690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/683690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/1"}}},"commit":{"parents":[{"commit":"96fa23758dfcd4cf7379bf1b229fd392c9c24bd7","subject":"Add build_active_retries and port_detach_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/96fa23758dfcd4cf7379bf1b229fd392c9c24bd7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 18:38:00.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 18:38:58.000000000","tz":-420},"subject":"Add parameters for amphora log offloading","message":"Add parameters for amphora log offloading\n\nThis patch adds the Octavia configuration parameters for amphora log\noffloading.\n\nChange-Id: I784c58d6898e843daa153aebb0c89482a7bb6050\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/726893199449ba830dc8904f6d9b95a8161de79b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/726893199449ba830dc8904f6d9b95a8161de79b"}]},"branch":"refs/heads/master"},"f351c9cd7e72f61d3e6df38b556e6bce445e0040":{"kind":"REWORK","_number":2,"created":"2019-09-20 19:36:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/683690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/683690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/2"}}},"commit":{"parents":[{"commit":"96fa23758dfcd4cf7379bf1b229fd392c9c24bd7","subject":"Add build_active_retries and port_detach_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/96fa23758dfcd4cf7379bf1b229fd392c9c24bd7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 18:38:00.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 19:36:21.000000000","tz":-420},"subject":"Add parameters for amphora log offloading","message":"Add parameters for amphora log offloading\n\nThis patch adds the Octavia configuration parameters for amphora log\noffloading.\n\nChange-Id: I784c58d6898e843daa153aebb0c89482a7bb6050\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f351c9cd7e72f61d3e6df38b556e6bce445e0040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f351c9cd7e72f61d3e6df38b556e6bce445e0040"}]},"branch":"refs/heads/master"},"952e448d8fb141be1ccd50eb12ca05cc80a78f15":{"kind":"REWORK","_number":3,"created":"2019-10-09 18:38:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/683690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/683690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/683690/3"}}},"commit":{"parents":[{"commit":"8f1d89c9e136ce3e6ac45f8f441f8f36bef41012","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8f1d89c9e136ce3e6ac45f8f441f8f36bef41012"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 18:38:00.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-09 18:38:18.000000000","tz":-420},"subject":"Add parameters for amphora log offloading","message":"Add parameters for amphora log offloading\n\nThis patch adds the Octavia configuration parameters for amphora log\noffloading.\n\nChange-Id: I784c58d6898e843daa153aebb0c89482a7bb6050\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/952e448d8fb141be1ccd50eb12ca05cc80a78f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/952e448d8fb141be1ccd50eb12ca05cc80a78f15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
