)]}'
{"id":"openstack%2Fpuppet-octavia~683776","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fstein~I8be48a7bd07e0a00972b4bef268c4bd2badc6711","project":"openstack/puppet-octavia","branch":"stable/stein","topic":"bug/1838410-stable/stein","hashtags":[],"change_id":"I8be48a7bd07e0a00972b4bef268c4bd2badc6711","subject":"Add support for connection retry variables","status":"MERGED","created":"2019-09-21 10:37:01.000000000","updated":"2019-09-23 11:32:24.000000000","submitted":"2019-09-23 11:30:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683776-1569238255687-7d46a8a7","meta_rev_id":"734e6f67eaeffe2bb2efd6c4d485581d3ed00d02","_number":683776,"virtual_id_number":683776,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-09-23 11:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-23 10:57:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-23 10:57:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-09-21 10:37:01.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 10:37:01.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-09-23 10:57:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-23 11:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05aabc3b397d6376062042809922a39560ed6a13","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-21 10:37:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e463f1eab5b9f24891b19b2393b82cd1169ae473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 11:04:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8db0c6abcbfd4989b4a32a62a318815a : SUCCESS in 4m 27s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d1c8043477748998952ecb0554c8816 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/618e2215b43c464789e76a94c894a42e : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05d5698278704df8a9731d2f94195d06 : SUCCESS in 4m 56s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/9e569fee4c4b44bcb952e95ee1d77b44 : SUCCESS in 13m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fb59bc3a5fe94dc8b243db7c4c853f7d : SUCCESS in 14m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0baf092f045b456d8f3577a49a9ba6c2 : SUCCESS in 24m 41s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c869c3372864140a7d8a5acb8b15032 : FAILURE in 8m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c6a638c323848b383b85c0c3075aa55 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f77310c0784f31c052d2187035a37bb1480272b2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-23 10:57:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44334098785011cbd393e3e5159a18b4815e6f1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 10:57:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"142b253c20872669fe749833ad8b8a6d8c58ba04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:30:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5211f3fc93a647498814cdc304f257df : SUCCESS in 4m 21s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36eb451896b24dafad1d16f95206f87e : SUCCESS in 3m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87a89b36beba4cf594192c1a3155acb1 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97d1f3540f4241a59864c6379928c4af : SUCCESS in 6m 28s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/869d28c69418413f9f9272122c5e15d8 : SUCCESS in 20m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5097a1cc500f4e8ba1c8e91ba85469d7 : SUCCESS in 20m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8ceb8d06eeed4d6399fb3bf9794f7731 : SUCCESS in 26m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7f91d22abf04b19a57077db3ffbe333 : FAILURE in 9m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0430d716eaf4d8bad5136bc1e5dff57 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"57ab71f7dbcfeeb27d37937d1da11d76504a5d83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:30:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4dc48b9525d70c85c07efb83783be342c58075b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:32:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5111af169bf4198b602fbed9edf1e54 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8652716babb20258235544f3fd0591194fc5c0a4","revisions":{"8652716babb20258235544f3fd0591194fc5c0a4":{"kind":"REWORK","_number":1,"created":"2019-09-21 10:37:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/76/683776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/76/683776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/76/683776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/76/683776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/76/683776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/76/683776/1"}}},"commit":{"parents":[{"commit":"e871346bf0b7642e8e0bf25d4de6977ce14cfe5c","subject":"Deprecate spare_amphorae_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e871346bf0b7642e8e0bf25d4de6977ce14cfe5c"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-30 13:04:02.000000000","tz":-240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-21 10:37:01.000000000","tz":0},"subject":"Add support for connection retry variables","message":"Add support for connection retry variables\n\nAdds support for connection_max_retries and connection_retry_interval\nconfiguration.\n\nChange-Id: I8be48a7bd07e0a00972b4bef268c4bd2badc6711\nCloses-Bug: #1838410\n(cherry picked from commit 36ca9c36cbd414c056d87835a384e821f7b41fcd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8652716babb20258235544f3fd0591194fc5c0a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8652716babb20258235544f3fd0591194fc5c0a4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
