)]}'
{"id":"openstack%2Fpuppet-octavia~683778","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fqueens~Ifc2a55c005c08240c875d3191e1031f69c2341aa","project":"openstack/puppet-octavia","branch":"stable/queens","hashtags":[],"change_id":"Ifc2a55c005c08240c875d3191e1031f69c2341aa","subject":"Add octavia::controller::controller_ip_port_list","status":"MERGED","created":"2019-09-21 10:40:51.000000000","updated":"2019-09-23 11:39:39.000000000","submitted":"2019-09-23 11:37:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683778-1569238670199-1ec4dea2","meta_rev_id":"30a4a9592e96ef5d2fc31b4ac51b4c3127204989","_number":683778,"virtual_id_number":683778,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-09-23 11:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-23 10:57:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-23 10:57:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-09-21 10:40:51.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 10:40:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-09-23 10:57:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-23 11:37:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f161322abea3ad04e37db1383d35fc8e31d109d4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-21 10:40:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"b2cf48825a519a5eb197dc297fca2f764a52a979","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 11:14:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/5826763378c642a9b7c8b355be457275 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/11da7eb45ad24a76b78d4c2cb0e52180 : SUCCESS in 5m 22s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/64de74e6b3e14a8eac3d3d6823f2e0e0 : SUCCESS in 15m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/86da1d363eb24999982984c42ff8b397 : SUCCESS in 12m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/73d876661b164f489b8b4bba01885e90 : SUCCESS in 27m 12s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/54de850f54354459951d6b4239b6f34a : SUCCESS in 15m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0496a8115ddd4053865135c4ee38caf3 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"43b8a547e35c0a4a7cb6ac720d3f2e7d5bf844e2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-23 10:57:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"915de43f9d8109d6800c1e839381e994d78e5d02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 10:57:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c6d7a7452625b03e1be4e87ce915be7956e51e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/20971b77a42a4e15be735b1936caaeae : SUCCESS in 5m 25s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/2efe3afe1f57435787ab26d78958a06e : SUCCESS in 5m 18s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/ebd5f7ca7f6d4bb890064a651e1475ee : SUCCESS in 14m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8fbd6952b8fa4243ab8441de2e9f79d5 : SUCCESS in 20m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8bffd66446294cd183e8240c6b303980 : SUCCESS in 37m 40s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/96cfc96c73bc462c8d9e10e4c186e0f9 : SUCCESS in 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdfb7951b25b41c181005102b814be0c : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ccaa2b1a9c8121c744408b2875c7b9d3fe9f33db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"30a4a9592e96ef5d2fc31b4ac51b4c3127204989","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:39:39.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af3e9d39a71c41829640e2d33a326afb : FAILURE in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d050e0cef7bc037e6007e55a7d39e3c8319b0d24","revisions":{"d050e0cef7bc037e6007e55a7d39e3c8319b0d24":{"kind":"REWORK","_number":1,"created":"2019-09-21 10:40:51.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/78/683778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/78/683778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/78/683778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/78/683778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/78/683778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/78/683778/1"}}},"commit":{"parents":[{"commit":"b868fa4272a6f932f68075d7f41f06afe3aae415","subject":"Deprecate spare_amphorae_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b868fa4272a6f932f68075d7f41f06afe3aae415"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-11-07 13:53:13.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-21 10:40:51.000000000","tz":0},"subject":"Add octavia::controller::controller_ip_port_list","message":"Add octavia::controller::controller_ip_port_list\n\nThis patch adds the octavia::controller::controller_ip_port_list\nparameter that sets the [health_manager]/controller_ip_port_list\nconfiguration option.\n\nChange-Id: Ifc2a55c005c08240c875d3191e1031f69c2341aa\n(cherry picked from commit 822bac93676f99a90ffa199f5d9f1fc0ee6937d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d050e0cef7bc037e6007e55a7d39e3c8319b0d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d050e0cef7bc037e6007e55a7d39e3c8319b0d24"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
