)]}'
{"id":"openstack%2Fpuppet-octavia~683780","triplet_id":"openstack%2Fpuppet-octavia~stable%2Frocky~I5e2d2db8e7113e0c50b7cbd6a7ed383492477180","project":"openstack/puppet-octavia","branch":"stable/rocky","hashtags":[],"change_id":"I5e2d2db8e7113e0c50b7cbd6a7ed383492477180","subject":"Add build_active_retries and port_detach_timeout","status":"MERGED","created":"2019-09-21 10:46:08.000000000","updated":"2019-09-23 11:40:38.000000000","submitted":"2019-09-23 11:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683780-1569238671636-91dfc13c","meta_rev_id":"1cb44ed6aa5b1d2010dc58eb8a3e5a478817ba6e","_number":683780,"virtual_id_number":683780,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-23 11:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-23 10:57:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-23 10:57:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 10:57:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-23 11:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d6d8e443c9525949c4a46ac63d8fef76ca66537","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-21 10:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"006c32f63e2670dc515bd9398b67bc9c475c8bdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 11:08:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/56db81086991481da8ba5d80690b9bfc : SUCCESS in 4m 48s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/d8addb6d918f408cb126bf4ee3935908 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/72b208ec33da4fd3951de2009ff9d06f : SUCCESS in 4m 41s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/11976e3b63ed4387a90ca23cf35ca6f9 : SUCCESS in 12m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6e802e052c0d4b71999147b404b77e20 : SUCCESS in 20m 19s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b8e4a1e935d47f682ab9fa5976a6fde : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/809aa167a581408ca4cee785de354f8e : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"89918c28d325bc2013835ba9c6559605bec63fa4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-23 10:57:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba036b5c2d14cce5e1eb5d64ea0751d3f675dd37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 10:58:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3e887914d8bb8d26af71a7004d7a49500a6421a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/1ef35f4046ce4fbf86d3c7a168d03bf2 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/688ced6ca9f14ca1b9b70151bd83c53b : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/155a3490297941dbb53e30e111ac7079 : SUCCESS in 3m 58s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/963717d773d54c888ad603349261e77b : SUCCESS in 20m 07s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9514bd47ddce41dabb91439b38ab24b2 : SUCCESS in 24m 58s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63a35721dc0946f3b0da31be66b7d9eb : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7170207c75a749efaf73a4783490fb8e : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"92c7d0b9705d5c47a737fe8b4a3cb4df1d47a49f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1cb44ed6aa5b1d2010dc58eb8a3e5a478817ba6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:40:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0da8f9db88a5421ca142167b3104e6f9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe115ce1d1c740c18fe1e0cae231ef951fc72e11","revisions":{"fe115ce1d1c740c18fe1e0cae231ef951fc72e11":{"kind":"REWORK","_number":1,"created":"2019-09-21 10:46:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/80/683780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/683780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/683780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/683780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/683780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/683780/1"}}},"commit":{"parents":[{"commit":"90cf0b562845db8c7d480674b2ebd7b5ef98737b","subject":"Add support for connection retry variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/90cf0b562845db8c7d480674b2ebd7b5ef98737b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 12:57:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-21 10:45:59.000000000","tz":120},"subject":"Add build_active_retries and port_detach_timeout","message":"Add build_active_retries and port_detach_timeout\n\nParameter build_active_retries sets retry threshold for waiting for a\nbuild slot for an amphorae.\n\nParameter port_detach_timeout sets seconds to wait for a port to detach\nfrom an amphora\n\nChange-Id: I5e2d2db8e7113e0c50b7cbd6a7ed383492477180\n(cherry picked from commit 96fa23758dfcd4cf7379bf1b229fd392c9c24bd7)\n(cherry picked from commit e925e9143fc3f6cbb0678b053cc2a0c11f04e66e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/fe115ce1d1c740c18fe1e0cae231ef951fc72e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/fe115ce1d1c740c18fe1e0cae231ef951fc72e11"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
