)]}'
{"id":"openstack%2Fpuppet-octavia~683781","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fqueens~I8be48a7bd07e0a00972b4bef268c4bd2badc6711","project":"openstack/puppet-octavia","branch":"stable/queens","hashtags":[],"change_id":"I8be48a7bd07e0a00972b4bef268c4bd2badc6711","subject":"Add support for connection retry variables","status":"MERGED","created":"2019-09-21 10:47:42.000000000","updated":"2019-09-23 11:37:52.000000000","submitted":"2019-09-23 11:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683781-1569238672296-5622d755","meta_rev_id":"128e7afbf01ca3b553f81389425b893879a7c70b","_number":683781,"virtual_id_number":683781,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-09-23 11:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-23 10:57:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-23 10:57:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-09-21 10:47:42.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 10:47:42.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-09-23 10:57:52.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-23 11:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd1c10269392c9072edaf29deeb90f599db9d31a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-21 10:47:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c433f68da0ed52024f6987091aae955f598e216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 11:17:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/ae363d8302324cc29daa1152e2119c7e : SUCCESS in 4m 34s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/8498fea73d1544b6bd2ac31a51387d7b : SUCCESS in 4m 14s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/f2b81ca374004459b8f6e7fd3838fbe0 : SUCCESS in 18m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1e42821cde844275965f7fbc88ecd292 : SUCCESS in 19m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7504d4df86f7421285f381ce9a3cd896 : SUCCESS in 22m 15s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/2a5065d03f804946a438035cdd60b573 : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c313793857487196c0eda5c6dd8f49 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"227add2e22567296b60673980c100723cc7da777","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-23 10:57:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"827e9160c9c37de787da3502cc39a5822d884837","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 10:58:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6d890e1dc2eefbc93270de1ea0cdbc1d78600e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/a113cbcec54f4593bb03d16698d51aaf : SUCCESS in 5m 21s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/2949e45877744c2183422fc5be8abcfa : SUCCESS in 6m 42s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/3f9b9ba00bc14073a32e5611ebc88bd2 : SUCCESS in 13m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/47b7b9defb8f4d14b9002ba084b13a93 : SUCCESS in 17m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1f3958363c87425b8e2468bac6251f6c : SUCCESS in 30m 35s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/365e05ca838c497cb0c26f2a9331b0aa : SUCCESS in 20m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc8039a252b648ccb87b5a28831cb3a7 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"128e7afbf01ca3b553f81389425b893879a7c70b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f3c127808eb431bb1349a07b27d654e4460dd9f","revisions":{"5f3c127808eb431bb1349a07b27d654e4460dd9f":{"kind":"REWORK","_number":1,"created":"2019-09-21 10:47:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/683781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/81/683781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/81/683781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/81/683781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/81/683781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/81/683781/1"}}},"commit":{"parents":[{"commit":"d050e0cef7bc037e6007e55a7d39e3c8319b0d24","subject":"Add octavia::controller::controller_ip_port_list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d050e0cef7bc037e6007e55a7d39e3c8319b0d24"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-30 13:04:02.000000000","tz":-240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-21 10:47:28.000000000","tz":120},"subject":"Add support for connection retry variables","message":"Add support for connection retry variables\n\nAdds support for connection_max_retries and connection_retry_interval\nconfiguration.\n\nConflicts:\n    manifests/controller.pp\n    spec/classes/octavia_controller_spec.rb\n\nChange-Id: I8be48a7bd07e0a00972b4bef268c4bd2badc6711\nCloses-Bug: #1838410\n(cherry picked from commit 36ca9c36cbd414c056d87835a384e821f7b41fcd)\n(cherry picked from commit 8652716babb20258235544f3fd0591194fc5c0a4)\n(cherry picked from commit 90cf0b562845db8c7d480674b2ebd7b5ef98737b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f3c127808eb431bb1349a07b27d654e4460dd9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f3c127808eb431bb1349a07b27d654e4460dd9f"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
