)]}'
{"id":"openstack%2Fpuppet-octavia~683782","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fqueens~I5e2d2db8e7113e0c50b7cbd6a7ed383492477180","project":"openstack/puppet-octavia","branch":"stable/queens","hashtags":[],"change_id":"I5e2d2db8e7113e0c50b7cbd6a7ed383492477180","subject":"Add build_active_retries and port_detach_timeout","status":"MERGED","created":"2019-09-21 10:48:10.000000000","updated":"2019-09-23 11:40:34.000000000","submitted":"2019-09-23 11:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683782-1569238672976-f5fe0389","meta_rev_id":"65f4e94836476cd06b8a6807b076d0c8f7ae582f","_number":683782,"virtual_id_number":683782,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-23 11:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-23 10:57:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-23 10:57:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 10:57:57.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-23 11:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66a6f80e7669fb0aabf0e509ba70f1ad8d9076d6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-21 10:48:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aabcac8f83c1e25a891ec1750f41b452e76214cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 11:23:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/5b27cb8c8cb84ba0b058a5c006438266 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/97ae3511b9674bf78524dec34cefd970 : SUCCESS in 5m 10s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/708d8f597738409abbbc8b70c85ecdd4 : SUCCESS in 12m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b7ccc5e25b7842c48ae6d8f64cf02330 : SUCCESS in 11m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fca09cbc385945bdaf4fe6874191bdcb : SUCCESS in 33m 16s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/1943ddd4787742189fb404ccceecf230 : SUCCESS in 14m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07f08fb40a4147c1aa14adf7c2b08caf : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5f56cbf0d8a9bb10592d77da94fa34e2384abd4d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-23 10:57:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e0728021c594dd37a9330389abac2a57d47e9dd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 10:58:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a4afa5b7f129b34af1ab05d77ff16e77a644ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/c43099b7bf1348cc9f25fd1b784ffebe : SUCCESS in 6m 08s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/8ff5ab2bfb504216ae454f9dc48b0f6f : SUCCESS in 6m 23s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/6726999cb78640f5877dc798f2426ca1 : SUCCESS in 18m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/eeb8fbd8a9274393bb3c0234806e27ff : SUCCESS in 18m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/aa607bd245274248a2423e79fb4e7aba : SUCCESS in 32m 20s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/633849696ab7402aa4a16e6d7774c9b8 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c1e39b96554b5e87ee0e40218496d8 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7f6a5d0af425645cc03530f7c8a89fb13349122c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:37:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65f4e94836476cd06b8a6807b076d0c8f7ae582f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:40:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/657c857286cf48589b5957f48dacaccc : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3acf4ef0ef1135c7df66ce39d56d873df7c96330","revisions":{"3acf4ef0ef1135c7df66ce39d56d873df7c96330":{"kind":"REWORK","_number":1,"created":"2019-09-21 10:48:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/82/683782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/82/683782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/683782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/683782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/683782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/82/683782/1"}}},"commit":{"parents":[{"commit":"5f3c127808eb431bb1349a07b27d654e4460dd9f","subject":"Add support for connection retry variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f3c127808eb431bb1349a07b27d654e4460dd9f"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 12:57:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-21 10:47:59.000000000","tz":120},"subject":"Add build_active_retries and port_detach_timeout","message":"Add build_active_retries and port_detach_timeout\n\nParameter build_active_retries sets retry threshold for waiting for a\nbuild slot for an amphorae.\n\nParameter port_detach_timeout sets seconds to wait for a port to detach\nfrom an amphora\n\nChange-Id: I5e2d2db8e7113e0c50b7cbd6a7ed383492477180\n(cherry picked from commit 96fa23758dfcd4cf7379bf1b229fd392c9c24bd7)\n(cherry picked from commit e925e9143fc3f6cbb0678b053cc2a0c11f04e66e)\n(cherry picked from commit fe115ce1d1c740c18fe1e0cae231ef951fc72e11)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3acf4ef0ef1135c7df66ce39d56d873df7c96330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3acf4ef0ef1135c7df66ce39d56d873df7c96330"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
