)]}'
{"id":"openstack%2Fpuppet-octavia~688110","triplet_id":"openstack%2Fpuppet-octavia~master~I26ad5d41239b6e770e975dc6f78fc46a87803220","project":"openstack/puppet-octavia","branch":"master","topic":"actstby","hashtags":[],"change_id":"I26ad5d41239b6e770e975dc6f78fc46a87803220","subject":"Add keepalivd vrrp parameters","status":"MERGED","created":"2019-10-11 12:04:18.000000000","updated":"2019-10-12 01:05:12.000000000","submitted":"2019-10-12 01:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688110-1570842249324-9904e72e","meta_rev_id":"4b72a0d3141031332f2bc0d6806f835747d397c5","_number":688110,"virtual_id_number":688110,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-12 01:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-10-12 00:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-10-11 22:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-10-12 00:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 12:05:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-11 12:05:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-10-11 12:05:18.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-10-11 22:17:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-12 00:38:33.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-10-12 01:04:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46a704b7f92dd73710491dc1ca942305615bf970","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-11 12:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb513770741af7e65dcaf45d0b040c6dbc1d6176","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-11 12:06:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6bb4bdd331acf8b3ac1b8812ad99682ce653b0b7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-11 12:24:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"611d8748a20a828e6a2869c2139490fd542b1380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 13:01:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd637c3b199d4951838fcac7fa9f63a1 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43ae1bbce1d54f02a434490ca1424101 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95f2c663511c4ed396aedfdad50e99f5 : SUCCESS in 3m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a6d098e09fb14e1bbfd2ce1fad6f521b : FAILURE in 10m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f577f2c8df0443b384a09839cabdbd02 : FAILURE in 9m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db100d5a4a4e4114ae4c4604faac3f09 : FAILURE in 9m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/057a57154c7c4372a36ca656a2e70414 : SUCCESS in 33m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cb6c8248a5a4308bab0b29f5ad7016c : FAILURE in 12m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c88454668e440f3afa5deed8270e624 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e95aa7ae15daaee1b303e1210aa46c1c8722060b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-11 13:17:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d7ec27876b9d2a25987c58a94be56ca212b541e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 13:46:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9027fa3a2c054b718ba5f8c4df96186d : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7eae265ad7ea4ca08f75ae90ea0634c1 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce413a0e65a84e1e9321bb3c39328362 : SUCCESS in 6m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2b017948b19c47f28f214dd15f8f00ee : SUCCESS in 18m 49s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cd31dffb87e46df820a2ac236c061a8 : SUCCESS in 16m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ffca4b9179141ac81f3c9c8a11ef8c9 : SUCCESS in 11m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3468097623484fc689f9fccf41384a81 : SUCCESS in 27m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d99279ca75504f52b28a3e6462f7001e : FAILURE in 10m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a3fc96bcc764cc1b7d6a2367f566002 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":4},{"id":"563e4d4e1f63b55c78d54968e9b96ba6de705b19","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-11 22:17:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3a51487b51d89b9eae35b4fa2cf661b6b2805093","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-12 00:38:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"975849962a180da5844ee53ceef28307c37d8c1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 00:38:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e058d624ee2899199dd9b0a7852af47ed9719d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 01:04:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c5ce401e22348bf98e732f7a09743a7 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4a1457d0e2848d195bb80ff932465a1 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d678c6dae9d9442eb84369f13e4e02fd : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/364d7f488ee54fa2a9544c9118d1d48f : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8c568d9ebbd4557b323ff607d078511 : SUCCESS in 11m 01s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/488a0c69d02149f884edb745e6479c0a : SUCCESS in 20m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaa7ad574a2546c0b9e5503d9f9ac407 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a20d5fcd1299f90a89f26a68f0d294833b84a97f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 01:04:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2c58e72dc13b69f1dc57fa2e6c65d17ef408d003","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 01:05:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1b712b8c49d4f4d8e784d9821fc7dd6 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8fd7b2e0ff231a602167af98f8fe7729c872b319","revisions":{"2f60052d82a853fa3404636fcde454ebec705bd0":{"kind":"REWORK","_number":1,"created":"2019-10-11 12:04:18.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/10/688110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/10/688110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/1"}}},"commit":{"parents":[{"commit":"952e448d8fb141be1ccd50eb12ca05cc80a78f15","subject":"Add parameters for amphora log offloading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/952e448d8fb141be1ccd50eb12ca05cc80a78f15"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 12:03:37.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2f60052d82a853fa3404636fcde454ebec705bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2f60052d82a853fa3404636fcde454ebec705bd0"}]},"branch":"refs/heads/master"},"427626897ec4fa0a6876f7f4868bbf14a8971e25":{"kind":"REWORK","_number":2,"created":"2019-10-11 12:06:44.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/10/688110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/10/688110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/2"}}},"commit":{"parents":[{"commit":"952e448d8fb141be1ccd50eb12ca05cc80a78f15","subject":"Add parameters for amphora log offloading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/952e448d8fb141be1ccd50eb12ca05cc80a78f15"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 12:06:36.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/427626897ec4fa0a6876f7f4868bbf14a8971e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/427626897ec4fa0a6876f7f4868bbf14a8971e25"}]},"branch":"refs/heads/master"},"3ffeaef12e171cab903f078e316801085683072d":{"kind":"REWORK","_number":3,"created":"2019-10-11 12:24:07.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/10/688110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/10/688110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/3"}}},"commit":{"parents":[{"commit":"952e448d8fb141be1ccd50eb12ca05cc80a78f15","subject":"Add parameters for amphora log offloading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/952e448d8fb141be1ccd50eb12ca05cc80a78f15"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 12:23:59.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3ffeaef12e171cab903f078e316801085683072d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3ffeaef12e171cab903f078e316801085683072d"}]},"branch":"refs/heads/master"},"8fd7b2e0ff231a602167af98f8fe7729c872b319":{"kind":"REWORK","_number":4,"created":"2019-10-11 13:17:51.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/10/688110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/10/688110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/10/688110/4"}}},"commit":{"parents":[{"commit":"952e448d8fb141be1ccd50eb12ca05cc80a78f15","subject":"Add parameters for amphora log offloading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/952e448d8fb141be1ccd50eb12ca05cc80a78f15"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 13:17:28.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8fd7b2e0ff231a602167af98f8fe7729c872b319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8fd7b2e0ff231a602167af98f8fe7729c872b319"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
