)]}'
{"id":"openstack%2Fpuppet-octavia~688279","triplet_id":"openstack%2Fpuppet-octavia~stable%2Ftrain~I26ad5d41239b6e770e975dc6f78fc46a87803220","project":"openstack/puppet-octavia","branch":"stable/train","hashtags":[],"change_id":"I26ad5d41239b6e770e975dc6f78fc46a87803220","subject":"Add keepalivd vrrp parameters","status":"MERGED","created":"2019-10-12 18:02:58.000000000","updated":"2019-10-17 12:54:47.000000000","submitted":"2019-10-17 12:53:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688279-1571316815642-9e252c24","meta_rev_id":"f9f1b7db8c068801dc089ba1cbc527cc0e5953b2","_number":688279,"virtual_id_number":688279,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-10-15 21:37:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 12:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-17 12:27:35.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-17 12:27:35.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 12:27:35.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-17 12:53:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db6171d0ef75deaa09ad8a070afaebd34de750c3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 18:02:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ca11be35d8cd190bda1bbdcb474a33c05181e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 18:29:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fca45ae5796d4dd98a1678027fad55c0 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c9c5d89bd234decaeb3b1e6a3cbf006 : SUCCESS in 3m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/01f36c6882c14309b2c6f9d3891184d1 : FAILURE in 7m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/77ba9c1708914800bebacbe99de2fbf2 : SUCCESS in 24m 56s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdd2efa64aad45b4bfe3bf560bbbf8c9 : FAILURE in 8m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/698f0f89308e43daa496936d9de3ec14 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"05d8e24e91edfa9dfc850506b5542e75cb00bcb3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-13 19:18:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32f512fe58b0270dcb1a78fc0fdbf09849eb27ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 19:38:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1bf5c78c20249b2adf6395d2e5299bd : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/902cb14118344e1096436c3dab674ce6 : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9d6f3ee41e094d9b802cf69e7320befe : FAILURE in 9m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b47ef1b0887241d9a7b2ee97e4be1cbc : FAILURE in 9m 26s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43de4a8ccf424f4b9c46305945f39310 : FAILURE in 7m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bba3db5283c4ab1b14b000a4a01adbf : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0c1bf3c8ac4d47ee13d83db445b2c21cfb4edd03","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-15 21:37:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3c97fc9bfdfb88037b4ae61fb75804aa24301a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:01:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ba9489e14f54f46aba3823c8d172782 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad6806cf3ecc4d7f9afddde02c08095a : SUCCESS in 3m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/816e1ae83c9243519b0c7683cff5a2d2 : SUCCESS in 9m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/39d11e981f5a438f9e65ebfbb1463610 : SUCCESS in 22m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0563698f22af4a42bb07d49eddf2c711 : FAILURE in 6m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64e0ed4398584d68b86116170bb5e433 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b3e1e07b7130d1a9147843f5cc5ec88a4244da64","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-17 12:27:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b46bd71180f1c8215d016bc6e3a0e9c5db0c14e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:27:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60b0b8877f58efc427b87c897741bc666ed1f276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:53:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adfc67c87cf545f8bc8647b77ed54873 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f133e6e978b44f6ba48cd91a022787a : SUCCESS in 3m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/54e5bb3cb9fe48b28247864288368c47 : SUCCESS in 17m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f3eb8aace478481c85a1cebc2330b3ee : SUCCESS in 20m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1c009346a00416c93910b269931ff87 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"807c1fca05cef5292f8735c9494a8a09c670ff47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:53:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f9f1b7db8c068801dc089ba1cbc527cc0e5953b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:54:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0799adff2da245199c39c16d8feec975 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b659647dbd21a48560750e611a9b7646f39f48be","revisions":{"f9ec3e241043748698b4c5a0aa023a16bd4d615f":{"kind":"REWORK","_number":1,"created":"2019-10-12 18:02:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/79/688279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/688279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/1"}}},"commit":{"parents":[{"commit":"b12f610f3a758aadea37f00f3c517ebf6bd49921","subject":"Switch to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b12f610f3a758aadea37f00f3c517ebf6bd49921"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-12 18:02:31.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n(cherry picked from commit 8fd7b2e0ff231a602167af98f8fe7729c872b319)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f9ec3e241043748698b4c5a0aa023a16bd4d615f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f9ec3e241043748698b4c5a0aa023a16bd4d615f"}]},"branch":"refs/heads/stable/train"},"b659647dbd21a48560750e611a9b7646f39f48be":{"kind":"REWORK","_number":2,"created":"2019-10-13 19:18:32.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/79/688279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/688279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/688279/2"}}},"commit":{"parents":[{"commit":"b12f610f3a758aadea37f00f3c517ebf6bd49921","subject":"Switch to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b12f610f3a758aadea37f00f3c517ebf6bd49921"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-13 19:17:31.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n(cherry picked from commit 8fd7b2e0ff231a602167af98f8fe7729c872b319)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b659647dbd21a48560750e611a9b7646f39f48be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b659647dbd21a48560750e611a9b7646f39f48be"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
