)]}'
{"id":"openstack%2Fpuppet-octavia~688280","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fstein~I26ad5d41239b6e770e975dc6f78fc46a87803220","project":"openstack/puppet-octavia","branch":"stable/stein","hashtags":[],"change_id":"I26ad5d41239b6e770e975dc6f78fc46a87803220","subject":"Add keepalivd vrrp parameters","status":"MERGED","created":"2019-10-12 18:03:39.000000000","updated":"2019-10-21 10:29:14.000000000","submitted":"2019-10-21 10:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688280-1571653643259-cd6aae0a","meta_rev_id":"25720e76b6fae453799f778d4e9581058acfdc1d","_number":688280,"virtual_id_number":688280,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 10:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-21 09:55:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-18 01:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-21 09:55:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 01:12:09.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-10-21 09:55:56.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-21 10:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b9e1731bc572f8ea8937914bfdf25df6f2c52ec","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 18:03:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"476498e0278c9bbf360e81a50524c1f925c08445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 18:30:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5913a847098c4bc9b30c89e7d95fc578 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/220f9ec4920f4d769bdc3254749c37b3 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7eb8e734e7e745d2a6de37cb90d0e689 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48cbae34dd1d4f1883ad19d903d8d195 : SUCCESS in 3m 49s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/f2745cae38cd4f45a5229a7a4f2437a3 : FAILURE in 7m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3bba06144d7e4aa1b402dcf41303db8e : FAILURE in 7m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/070e0bed823c47979857b4ef4fe54c33 : SUCCESS in 25m 14s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88cf53c6b271458eb5514bc0c968c446 : FAILURE in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cb602bba0ae44b3a927b959c305737d : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7de134035c8106622907d6bde513414669eb633a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-17 12:58:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"840840a2f6eceeced5a1e6a64c37f46ca3763031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 13:28:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e4c9d32100a4313b7df37d404657b67 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fdb483d817e4b548b829134492ce8b4 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c7fd63c1a8148ee8d62189e70dfdeb5 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/397d29127a1d4b6da6f391c985904351 : SUCCESS in 5m 10s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/47c9a1ab579842f09e96dca9acf9d45b : SUCCESS in 13m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/eaa43fb5282b41fc9fdf03934cf36240 : SUCCESS in 12m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/864799886fee4ed8b463e938b4296f94 : SUCCESS in 27m 03s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4778dcc3c790463a9e161ad10a0714aa : FAILURE in 9m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bce2dc2bf7240baa3867c93b6263bf6 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b5acef1aa69d7cbf0c3b7603cd2821310f91bc10","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-18 01:12:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35f96708d52ca7095b5ccb45cea6ca2d93568fd1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-21 09:55:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e2835a312b40572923ea57f68995600afd0a2f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 09:56:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d52ff425166978cf53b202cf10435ebc494c8db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:27:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c60d96ddf8a5465a975b7027d3a8582b : SUCCESS in 4m 25s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0510ce657f7341099155cfb5f9b334ae : SUCCESS in 5m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99e3b549a883476981915ee022964695 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b09c226e49b4f8c9b7fd85ae97962b4 : SUCCESS in 4m 46s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/944128017de94cd79dc9dcbd1339c501 : SUCCESS in 18m 30s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e56e8bf69442499cb47f8671cbbc2f35 : SUCCESS in 14m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/68aeead080c242309d17934d90c1f916 : SUCCESS in 29m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e7c4245c5794ea79f5063fab38fd9cd : FAILURE in 13m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f5e2e8930ed40248e679766823f8a36 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6f048cffd5fcf5ef23cb12c49ea7f98dafc0a735","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:27:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"25720e76b6fae453799f778d4e9581058acfdc1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:29:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca8dc9bae19d4d0ea8d8421d5d407dd5 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3de4adbe19bf79e46b4ffc8fa3aafda08684f951","revisions":{"511f5cf4cd4b3a646380b8600db079363484b0e7":{"kind":"REWORK","_number":1,"created":"2019-10-12 18:03:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/80/688280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/688280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/1"}}},"commit":{"parents":[{"commit":"cad740722f7da8132456fb4538a03291a53a5d71","subject":"Add the \"connection_logging\" parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cad740722f7da8132456fb4538a03291a53a5d71"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-12 18:03:28.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n(cherry picked from commit 8fd7b2e0ff231a602167af98f8fe7729c872b319)\n(cherry picked from commit f9ec3e241043748698b4c5a0aa023a16bd4d615f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/511f5cf4cd4b3a646380b8600db079363484b0e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/511f5cf4cd4b3a646380b8600db079363484b0e7"}]},"branch":"refs/heads/stable/stein"},"3de4adbe19bf79e46b4ffc8fa3aafda08684f951":{"kind":"REWORK","_number":2,"created":"2019-10-17 12:58:14.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/80/688280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/688280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/688280/2"}}},"commit":{"parents":[{"commit":"cad740722f7da8132456fb4538a03291a53a5d71","subject":"Add the \"connection_logging\" parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cad740722f7da8132456fb4538a03291a53a5d71"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-17 12:57:45.000000000","tz":60},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n(cherry picked from commit 8fd7b2e0ff231a602167af98f8fe7729c872b319)\n(cherry picked from commit b659647dbd21a48560750e611a9b7646f39f48be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3de4adbe19bf79e46b4ffc8fa3aafda08684f951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3de4adbe19bf79e46b4ffc8fa3aafda08684f951"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
