)]}'
{"id":"openstack%2Fpuppet-octavia~688282","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fqueens~I26ad5d41239b6e770e975dc6f78fc46a87803220","project":"openstack/puppet-octavia","branch":"stable/queens","hashtags":[],"change_id":"I26ad5d41239b6e770e975dc6f78fc46a87803220","subject":"Add keepalivd vrrp parameters","status":"MERGED","created":"2019-10-12 18:05:25.000000000","updated":"2019-10-21 10:29:03.000000000","submitted":"2019-10-21 10:27:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688282-1571653644431-f62c7b6a","meta_rev_id":"bcc1e2c09ca2fa4edab99fee7616e4d36b5de755","_number":688282,"virtual_id_number":688282,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 10:27:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-21 09:56:01.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-18 01:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-21 09:56:01.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 01:12:22.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-10-21 09:56:01.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-21 10:27:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba693be47b2b95e6c9b6ed7afadce6e4c61c77db","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 18:05:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1c806cf1e5c3aa2a0bcac7b5f450690f326d59","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-12 18:06:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f8ef30bcce66d5a6f4ee19767806eb794eae57f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 18:26:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/aefd9d06889a48818842b40b37bca1de : SUCCESS in 4m 02s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/b64ad1d594d64b1da455f4cf8ed7089f : SUCCESS in 3m 41s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/1e2369a61ab447509f0224b0e4594338 : FAILURE in 7m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/365e700406884089a48098a7d95789a5 : FAILURE in 13m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/274663564ae84524be25cecab5f11eb4 : SUCCESS in 18m 18s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/fd91c41e04b3464ab5b748753f47103c : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fbdb6604d184f3bb282cc0120b7df7d : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bf5d1fc65f33e29d5919a5e50e233fd11f03236e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-17 13:00:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2d39cf5012a48068d69b6ca9a12066da53fcecf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 13:28:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/e4e52e61ea994eef948713fdbb18b21f : SUCCESS in 7m 32s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/7282de805626434ebffce0f0fe6c5f17 : SUCCESS in 4m 13s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/afe0cb8ee356449880ed79ec4836d419 : SUCCESS in 18m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/383f2deb5b2346ff9cb318715b618d96 : SUCCESS in 17m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/20c5c12985ac42dda1cf63d2ed776d12 : SUCCESS in 22m 35s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/f1d03061c896451ca2744cb45c122d85 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5335e840824e3e932ee687e38b0f10 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3d02514f2079b6d7747823c74ddba00ff2aee9d2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-18 01:12:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ce8f75b5f33c4adffa6f12ceec7f4b22bc478fb4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-21 09:56:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"51e6015b6ed046f44db3d4069c52d49fe5e8bc98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 09:56:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3ebf479d86462a47c4f21179d2be59d5b5f45a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:27:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/36b8c4badf9b4b76a7800d78bb40e358 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/5239cd370b714e8ea887872dbcfebdd3 : SUCCESS in 5m 26s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/516c88cd16ff41d1b580357b1726271f : SUCCESS in 16m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a39266930c20449895af9e942fc36807 : SUCCESS in 16m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/25832425b6714825b4c3a96b35dba509 : SUCCESS in 22m 14s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/e1a567e8ec744cefb8a78206b0624202 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b08acdd88b3489a9d616929da9c1c60 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"76dc8d6d1d5ba254bf2defe18c3b98e4329a86d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:27:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bcc1e2c09ca2fa4edab99fee7616e4d36b5de755","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:29:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78e08db8006f47f1a7e3d239188756f9 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"251c78b7ae9ded1c27f228a9ea9952988e3fe05b","revisions":{"dacc76049cacbf3c2f0ada1410c1e451ae7a8806":{"kind":"REWORK","_number":1,"created":"2019-10-12 18:05:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/82/688282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/82/688282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/1"}}},"commit":{"parents":[{"commit":"a44decf5ba2ecb0fd402e5b9216db1b373a92389","subject":"Add the \"connection_logging\" parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a44decf5ba2ecb0fd402e5b9216db1b373a92389"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-12 18:05:17.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n(cherry picked from commit 8fd7b2e0ff231a602167af98f8fe7729c872b319)\n(cherry picked from commit f9ec3e241043748698b4c5a0aa023a16bd4d615f)\n(cherry picked from commit 511f5cf4cd4b3a646380b8600db079363484b0e7)\n(cherry picked from commit 1deb336080d65869cde9600f06aa09b6e7cd09bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dacc76049cacbf3c2f0ada1410c1e451ae7a8806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dacc76049cacbf3c2f0ada1410c1e451ae7a8806"}]},"branch":"refs/heads/stable/queens"},"41522cfabf237adcf2b780c9857b5f0e6802dd1f":{"kind":"REWORK","_number":2,"created":"2019-10-12 18:06:31.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/82/688282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/82/688282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/2"}}},"commit":{"parents":[{"commit":"a44decf5ba2ecb0fd402e5b9216db1b373a92389","subject":"Add the \"connection_logging\" parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a44decf5ba2ecb0fd402e5b9216db1b373a92389"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-12 18:06:24.000000000","tz":120},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n(cherry picked from commit 8fd7b2e0ff231a602167af98f8fe7729c872b319)\n(cherry picked from commit f9ec3e241043748698b4c5a0aa023a16bd4d615f)\n(cherry picked from commit 511f5cf4cd4b3a646380b8600db079363484b0e7)\n(cherry picked from commit 1deb336080d65869cde9600f06aa09b6e7cd09bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/41522cfabf237adcf2b780c9857b5f0e6802dd1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/41522cfabf237adcf2b780c9857b5f0e6802dd1f"}]},"branch":"refs/heads/stable/queens"},"251c78b7ae9ded1c27f228a9ea9952988e3fe05b":{"kind":"REWORK","_number":3,"created":"2019-10-17 13:00:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/82/688282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/82/688282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/82/688282/3"}}},"commit":{"parents":[{"commit":"a44decf5ba2ecb0fd402e5b9216db1b373a92389","subject":"Add the \"connection_logging\" parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a44decf5ba2ecb0fd402e5b9216db1b373a92389"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-11 08:16:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-10-17 13:00:41.000000000","tz":60},"subject":"Add keepalivd vrrp parameters","message":"Add keepalivd vrrp parameters\n\nChange-Id: I26ad5d41239b6e770e975dc6f78fc46a87803220\n(cherry picked from commit 8fd7b2e0ff231a602167af98f8fe7729c872b319)\n(cherry picked from commit b659647dbd21a48560750e611a9b7646f39f48be)\n(cherry picked from commit 3de4adbe19bf79e46b4ffc8fa3aafda08684f951)\n(cherry picked from commit f49cfd3bd0f3cd9347e9381012f72357b37b1468)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/251c78b7ae9ded1c27f228a9ea9952988e3fe05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/251c78b7ae9ded1c27f228a9ea9952988e3fe05b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
